)]}'
{"id":"openstack%2Fkayobe~721983","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Iabbe4c624b9ad02bb82c323070f99c16e5822966","project":"openstack/kayobe","branch":"stable/train","topic":"fix-remote-seed-vm-provisioning-stable/train","hashtags":[],"change_id":"Iabbe4c624b9ad02bb82c323070f99c16e5822966","subject":"Fix seed VM provisioning on a remote seed hypervisor","status":"MERGED","created":"2020-04-22 14:22:52.000000000","updated":"2020-04-23 11:39:15.000000000","submitted":"2020-04-23 11:35:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721983-1587641750726-cd7ad812","meta_rev_id":"b7d7f82ef3bd684156206338c57a34f49cf16eeb","_number":721983,"virtual_id_number":721983,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-23 11:35:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-23 09:21:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-22 16:41:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-23 09:21:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-22 16:41:36.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-23 11:35:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9169e89bdff313c9f99ca0f9f6797ed0d0219134","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-22 14:22:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"88f2e8113e767e1a22bdc02647379594e0677d65","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-22 16:41:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ad4b637f0447b721c14606241c5cb9886d3e6c0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-22 16:44:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99a08d7d86c34b8cb3ab87072b9fbc49 : SUCCESS in 6m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/767e33f292eb4fd0a197b4bb185f5be0 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d09c6767d73449aba4fb7d3e3176b2c : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3469932c8c2747879e803a017eefd4bb : SUCCESS in 6m 35s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0913823244e14f1685aae30792a556a6 : SUCCESS in 6m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93232d966c3f42b586987191a8eeccf3 : SUCCESS in 9m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0c8ed5a50bf494a957b05a395ac5dc3 : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0daa54726b91483dadfb9a05da91a498 : SUCCESS in 7m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b06ae1ae30954379be85f212432db9e1 : SUCCESS in 12m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/73229843c0f547c48666cca8621e6dee : SUCCESS in 32m 42s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/3704afafd7914cf9aaaf12cd84fde1a8 : SUCCESS in 1h 14m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4bc9af029d53425bbf140a2be49ec4e6 : SUCCESS in 1h 15m 47s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f7f1dd57ae574e3eb474f85ce399a9a7 : SUCCESS in 1h 59m 45s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a5f181535d3e44bbbcfa90beb5c4aab0 : SUCCESS in 24m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ba7fd76eaa784594b35c3f9c335f640f : SUCCESS in 19m 42s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/7e6cfb9827334210a4723be8723620c3 : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":1},{"id":"b5e8f85ce81f15b4cc35f3695cb12b9870771500","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 09:21:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"98a0a967d07d0d34700fa6e679ac2431663f947b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 09:21:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cfffb8fbc281570315f182339c9bf4e91e22073f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:35:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/76e30e53dbd04bce98c96dc194051d10 : SUCCESS in 10m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/28529619b3d84e45a876155b6c4b80f5 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bef59588752245ab8e42b417bdbd3662 : SUCCESS in 8m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c31578a817064a10ace8f5f188056ae3 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dffecac7b99e4d46ab733e7d2c3143a7 : SUCCESS in 8m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca865d15266440c19fd478af765b3b40 : SUCCESS in 5m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2f925b51bdc445dcb32d2bb94bd8851f : SUCCESS in 7m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b4b34f0d94de4bc8991eac7cebc9f9da : SUCCESS in 11m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a71fb99a3e7b4241919f880c5be413c3 : SUCCESS in 37m 44s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e53119412dde4fd9aa448b7513d36ff2 : SUCCESS in 1h 17m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e86aa6ae323745ee813e4f2134eaaa2d : SUCCESS in 1h 11m 13s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/295ff0b85b38425f88d77fd48d4598cb : SUCCESS in 2h 08m 55s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/165d72e92e15426b81d14e1b8b32978b : SUCCESS in 22m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1cb6b7f126c64e91afc7b2ed24d2a138 : SUCCESS in 19m 13s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/86b41d5c78404be9b3f93313b319067c : SUCCESS in 35m 46s","accounts_in_message":[],"_revision_number":1},{"id":"67e2bcee272a46ae5e16510732ad5f60bdd82038","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:35:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ab7d7b14cb2870a1556bbe0617060969194488ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:39:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42e5e45383604a0a96852a9484f2cf92 : SUCCESS in 2m 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca928a17b9c54d658dca37d56dde5813 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"41c528494aecea32f16e52df8be33c2e33ba34f0","revisions":{"41c528494aecea32f16e52df8be33c2e33ba34f0":{"kind":"REWORK","_number":1,"created":"2020-04-22 14:22:52.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/83/721983/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/721983/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/721983/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/721983/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/721983/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/721983/1"}}},"commit":{"parents":[{"commit":"23080784c44e3901c7835c56c0183c0200d6029f","subject":"Merge \"Prevent openrc files from using wrong OS_CACERT value\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23080784c44e3901c7835c56c0183c0200d6029f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-08 17:02:19.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-22 14:22:52.000000000","tz":0},"subject":"Fix seed VM provisioning on a remote seed hypervisor","message":"Fix seed VM provisioning on a remote seed hypervisor\n\nThe seed VM will fail to provision if the Ansible control host and the\nseed hypervisor are not the same hosts.\n\nThis is because Kayobe creates the seed-vm-user-data file on the\nseed-hypervisor host. It then invokes the jriguera.configdrive role\nwhich uses a copy task without remote_src, which fails to find the\nsource file locally on the Ansible control host.\n\nInstead we create a local temporary file for seed VM user data.\n\nChange-Id: Iabbe4c624b9ad02bb82c323070f99c16e5822966\nStory: 2007530\nTask: 39338\n(cherry picked from commit cc3d27e2e128f1203b7243ef1f7fd4ab207eb4ff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c528494aecea32f16e52df8be33c2e33ba34f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/41c528494aecea32f16e52df8be33c2e33ba34f0"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
