)]}'
{"id":"openstack%2Fkayobe~722236","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I33813f79984a46f1967ef3aee455dcfbe7eb93da","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I33813f79984a46f1967ef3aee455dcfbe7eb93da","subject":"Fix passwords.yml generation with vault encryption on Python 3","status":"MERGED","created":"2020-04-23 09:15:39.000000000","updated":"2020-04-24 15:00:48.000000000","submitted":"2020-04-24 14:55:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722236-1587740155032-653b66dd","meta_rev_id":"3be87e3dac3e62b819a76506bbd6414669baf170","_number":722236,"virtual_id_number":722236,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"date":"2020-04-24 10:58:17.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-24 14:55:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-04-23 12:27:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-23 09:51:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-04-23 12:27:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-23 09:51:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-23 12:27:46.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-24 14:55:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5c61b88c81c77abf23383c61144c78242cb82fbb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 09:15:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"48160700b6d4396dff90db8a388e64507ed9f3f9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-23 09:45:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01f1c98c1abc69e91d2a3c6abb6358dee9dae1aa","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 09:47:40.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2d32c65d785d36c6d3c070ac4ff25d61a388ef97","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 09:47:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b5fe47b304530915a5c2e9c84eca0fe8a47da6a4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-23 09:51:02.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"127827ddfc5bab4df09dc90515f9f1e7d3413a35","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 11:46:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5862976f4b474a3680a02ee45d8ba669 : SUCCESS in 7m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/34b88adf45e94f3d91474ee45575a04a : SUCCESS in 8m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ea2ea0cb5d3640be81e546d28fbb4529 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ffd6312d834e09971157cb5a792c07 : SUCCESS in 8m 59s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b2409e99090a4fb1b72e7a744cc31d5c : SUCCESS in 6m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bfba539a95246548bfaa94271fcac62 : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a3075fbbed14974b41bf718216b668c : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5856bb5a0f24412b9544ed22c266f73f : SUCCESS in 7m 48s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c66e9982107408a879c1d42b63d759a : SUCCESS in 12m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bc1f46e33fda4c8b94154dc8b6488dbb : SUCCESS in 35m 54s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/2040b0c132cb44c9a43412a98c517f35 : SUCCESS in 1h 11m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6a27942bf14a487b9aa933bc812dc3ed : SUCCESS in 1h 11m 12s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8e5d0c2df0f34d0d98fcc4175d843fcf : SUCCESS in 1h 54m 35s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/276434f32a3f4a82a43d9004b157384d : SUCCESS in 22m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/755b057d594a43d6be8d79487fb2ab40 : SUCCESS in 20m 44s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/81ecdb55fcc64975b94f3be495b20ac4 : SUCCESS in 31m 44s","accounts_in_message":[],"_revision_number":2},{"id":"0f9e0b3d3844b327a0616baf79559c3c46617673","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-04-23 12:27:46.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"904f68c4eb1db35a580943b8145a6248a15cc088","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 12:28:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d4e0849c923e1b089e445b02cb3907e88d1b35f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 13:49:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3af9ed9bfe5e41599b808c45fdb86c37 : SUCCESS in 6m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ae803ee42c4665b7ffcb5c83cfd752 : SUCCESS in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d56cb367de1482ca094a3ab3722cec5 : SUCCESS in 6m 57s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa91e8db99d44a5593ca806007a3c990 : SUCCESS in 7m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5070ebfcb66a49119444165161285bcc : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a805349ca25f478b87eafc5a5b1da8ca : SUCCESS in 5m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b18f047585c9466da46fc45e4a5e5a47 : FAILURE in 8m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9226bf501df945e9ba4f51d9b8d341a8 : FAILURE in 6m 26s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/461b6c6a7eff41859475ccc628d2916b : FAILURE in 9m 09s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/383f4145053a443795eea9f0baa2486c : FAILURE in 6m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/14f13ae0d951414db1ac76dc3ef33d95 : FAILURE in 7m 09s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/5024f18d8c8b46a1878f29698ae8f61d : FAILURE in 6m 16s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/46a2fd5c740a480cbd25942b590c43f4 : FAILURE in 8m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a93204e6a55542ad934956f5163bb80b : FAILURE in 8m 01s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/eefbf3cd837f415dafd2600b2245988f : FAILURE in 6m 44s","accounts_in_message":[],"_revision_number":2},{"id":"5ac292b1e9fd85e9b3170de107f41ec8253a2183","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 19:58:48.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"9e4c8811c9c3dd0ba34ecd54c8deb063bc71fb67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:50:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f23df55b844e437ebb0c9428bb08d4ad : SUCCESS in 7m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2886a9d5d83b48048011f8dd238e0f56 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ec75d5ffe40453095b31324584f47bd : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1dd9ff08324744828f2a8e7b268788fe : SUCCESS in 7m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35d0f93b872840398f0a89883d7f26fa : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dcdcd81eb2d485c8109c98f451d5d1b : SUCCESS in 8m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/412072a15da34b8c8cb03033ab7e1b52 : SUCCESS in 4m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f6c3b790f84f465c9e5198b391965394 : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/14c7194d59994a169a4074db0191c4eb : SUCCESS in 13m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8d2bf3385fac4c219022a01b184cce24 : SUCCESS in 36m 44s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/a695d04d87484aeb861eee73b672c6e3 : FAILURE in 32m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6598a3a2630c4f8293d362d7b0d9e956 : SUCCESS in 1h 04m 10s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/34c4a31d163d4611979b470a7285c8f4 : SUCCESS in 1h 45m 51s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/270bb0d1e1e24612ab39e20adaf984a0 : SUCCESS in 24m 07s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e4a2491d31b241dbb903cb85f1faa3b8 : SUCCESS in 18m 05s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8553cbc6691c4a8ea3baa90cb59d160d : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":2},{"id":"7a851f651b8e20fea7b23171b18f209e85828345","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-24 10:58:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"974be879948c38db07051a4ab458d23af66bf7da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 13:01:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af424b8c5e8a4eb9a0137cc10cf8677d : SUCCESS in 5m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/799df4b0cccb41ff8accc5b9e9ff1b78 : SUCCESS in 5m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3bfbd9f8d27941cc8cf2eeea4c44a255 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcfe393141a945a192624f27086c8e8b : SUCCESS in 6m 09s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35ff731ddb57464088efe76100c601c1 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5c1a4541934c15b40cbbe3a4ff1c79 : SUCCESS in 10m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8119372f29e2422db66aa397cb4df884 : SUCCESS in 5m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cc08133faf074b41acd78119e52ef1b3 : SUCCESS in 4m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0e00b2ac018b4bc5b562cc51c6ed56c4 : SUCCESS in 10m 08s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6d5460a779e14decb458b0777927b7ec : SUCCESS in 34m 55s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/ec56dd6ce54a4adf8870ac57f2278a40 : SUCCESS in 1h 10m 42s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/844cfab6971a4d8490a1b3543fa7b40e : SUCCESS in 1h 06m 55s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/069910c2cf284c8eb0bdf28104990941 : SUCCESS in 1h 58m 22s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/e2563df4235a439e8b75d6535682437b : SUCCESS in 24m 28s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4426886fb42e4bb4bd87967640811b53 : SUCCESS in 18m 03s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/51ee45d2d2ac487c85545a113bc0bf0e : SUCCESS in 43m 04s","accounts_in_message":[],"_revision_number":2},{"id":"1c63dfe948e25a3d0a065b265c874c0bbeb35808","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 13:01:44.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8f132967cb8c481954d22ed47ea961d7f93058eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:55:54.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3db070e791d14bb591a245c054ccaaf7 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91a722debd934a73ad3ec852a8abbee8 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a9e19b41d984cd4a9bb7cd20a616bbf : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ef90e627d1034fc1a7b1d9e8094e85f0 : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec1ceeade8cd4f9fa97c68c20a0beef6 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1c57c0520f549229fe100b94912a2f9 : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/851c4391f3ad45fc95e90342d8b257a1 : SUCCESS in 6m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/219fe90a0aff47b1a0f12f0c60aba4cc : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/705ac382634e4dcdb7b3258a20eb04ae : SUCCESS in 38m 54s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/263f09edc326441ba14ecec681aa1a90 : SUCCESS in 1h 13m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/820b1233580e452591e54dee99ff3a3a : SUCCESS in 1h 14m 31s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0c1ec0214f0d480189459d1392b3c926 : SUCCESS in 1h 52m 59s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/3e96d258b51c4f809835eb1e7584b99e : SUCCESS in 21m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/daa5c3004f7a4c20a7c02f6bb403d010 : SUCCESS in 18m 42s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8e633ab2f31647fb9c8b3c8f9bdb4daa : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":2},{"id":"63cbf07872effbafa05660c0821a54335c48ea3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 14:55:55.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"3be87e3dac3e62b819a76506bbd6414669baf170","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-24 15:00:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf59c7025ba84542a31093ca48e6a78b : SUCCESS in 2m 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2eef11ce4d224db6a41d5127342875ba : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f6a24dd35f7cd78e1bf29f8a14a178dfa081def2","revisions":{"28e7fe40591dbdfebaaf766e4941772b44b62ab4":{"kind":"REWORK","_number":1,"created":"2020-04-23 09:15:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/36/722236/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/722236/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/722236/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/722236/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/722236/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/722236/1"}}},"commit":{"parents":[{"commit":"23080784c44e3901c7835c56c0183c0200d6029f","subject":"Merge \"Prevent openrc files from using wrong OS_CACERT value\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23080784c44e3901c7835c56c0183c0200d6029f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-20 14:36:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-23 09:15:18.000000000","tz":60},"subject":"Fix passwords.yml generation with vault encryption on Python 3","message":"Fix passwords.yml generation with vault encryption on Python 3\n\nKayobe generates passwords.yml for Kolla Ansible, and can encrypt it\nusing the vault password. Previously this was failing on Python 3 due to\npassing a string to file.write() which expects bytes.\n\nThis change fixes the issue by encoding the password string passed to\nfile.write().\n\nThis allows us to run the ansible role tests under Python 3.\n\nChange-Id: I33813f79984a46f1967ef3aee455dcfbe7eb93da\nStory: 2006574\nTask: 39481\n(cherry picked from commit 8bed623571d8bc1695ac22a896e9efac2727f40d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28e7fe40591dbdfebaaf766e4941772b44b62ab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/28e7fe40591dbdfebaaf766e4941772b44b62ab4"}]},"branch":"refs/heads/stable/train"},"f6a24dd35f7cd78e1bf29f8a14a178dfa081def2":{"kind":"REWORK","_number":2,"created":"2020-04-23 09:47:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/36/722236/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/36/722236/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/722236/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/722236/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/36/722236/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/36/722236/2"}}},"commit":{"parents":[{"commit":"23080784c44e3901c7835c56c0183c0200d6029f","subject":"Merge \"Prevent openrc files from using wrong OS_CACERT value\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/23080784c44e3901c7835c56c0183c0200d6029f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-20 14:36:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-23 09:47:37.000000000","tz":0},"subject":"Fix passwords.yml generation with vault encryption on Python 3","message":"Fix passwords.yml generation with vault encryption on Python 3\n\nKayobe generates passwords.yml for Kolla Ansible, and can encrypt it\nusing the vault password. Previously this was failing on Python 3 due to\npassing a string to file.write() which expects bytes.\n\nThis change fixes the issue by encoding the password string passed to\nfile.write().\n\nThis allows us to run the ansible role tests under Python 3.\n\nChange-Id: I33813f79984a46f1967ef3aee455dcfbe7eb93da\nStory: 2006574\nTask: 39481\n(cherry picked from commit 8bed623571d8bc1695ac22a896e9efac2727f40d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a24dd35f7cd78e1bf29f8a14a178dfa081def2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f6a24dd35f7cd78e1bf29f8a14a178dfa081def2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
