)]}'
{"id":"openstack%2Fkayobe~722316","triplet_id":"openstack%2Fkayobe~master~Ia3c1f6423c32fa2580b57db32512a34ce35d7acc","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ia3c1f6423c32fa2580b57db32512a34ce35d7acc","subject":"Enable \u0027data\u0027 LVM group only for Docker devicemapper","status":"MERGED","created":"2020-04-23 13:21:23.000000000","updated":"2020-06-05 15:25:26.000000000","submitted":"2020-06-05 15:22:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"722316-1591370556977-c76c2b72","meta_rev_id":"8a1d5f26168c65405e9762fcee7396472ef0f1cd","_number":722316,"virtual_id_number":722316,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2020-04-23 19:57:44.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 15:22:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-05 09:38:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-05 13:12:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-05 13:12:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 09:38:05.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-05 13:12:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-05 15:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6f3b000ac222d13c22f6c313e1d59a8522a83b7d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 13:21:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9d07c4912e47fc73598074c0783f864553e89ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 13:37:09.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/62d60ce9f51442e2a727a9881cee0be1 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecea1c68cc054f4fa0fb8479b0b2535d : SUCCESS in 8m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebff6c8f11e34b358af203be7eb913fd : SUCCESS in 5m 47s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9f9cb7bc792845fc9c004a29aca77ff9 : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e619401fac5e4358ae8905120ca56880 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84fe4b2b849465dbacb1317d2330d44 : SUCCESS in 11m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6a92d961ec04ea28ecf749c2a1b6760 : SUCCESS in 5m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/51f2a1d2c4be4b0594c08a3e10fd6420 : FAILURE in 8m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/00786267b4764894b0850d7060f79a21 : FAILURE in 8m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fca7661b44164f0c97bd030f7199f196 : FAILURE in 9m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a50fb481dc3a495781d473becf9b84a0 : FAILURE in 8m 20s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fbb82d0217f94fcf9c49ac0b972a10b7 : FAILURE in 7m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/927f15ebeae447e887020f2f713abb8e : FAILURE in 7m 11s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/41111f3c356b4d84a037ec0581817522 : FAILURE in 8m 11s","accounts_in_message":[],"_revision_number":1},{"id":"0407902be8d0616a856fcb7791f58d7b83300f22","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-23 19:57:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"ca5463a3c6aacbe2bee0a3108375b74a23b4b3b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-23 21:51:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39791fdf28ba4b2d9d5660519100efa3 : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/152d67ffc15b4576a53582cf1fde4cdc : SUCCESS in 7m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/155708bf3d3c4b6d829f65d8ca4c8e6a : SUCCESS in 7m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3aa1d17777b84f7ba784235148d5c3f0 : SUCCESS in 6m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b962c4dbbca44ee3813c31c7eb53a230 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/910c2e5d9c5d46d4b1fdda984ff08d7c : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/66e07fb7a9d441049b718d38f873792e : SUCCESS in 4m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/437a4e868ff64cff9f883b06b960fd51 : SUCCESS in 6m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/11069fe79f6d41a08f59fbd65a133c8a : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8a5bed4060a14f35a4e6144bd27dc8c6 : SUCCESS in 29m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/34afad7089da4bc293bcae534560792b : SUCCESS in 1h 10m 08s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3fd671b1ff0c4876bd8c75b14a1ffaae : SUCCESS in 1h 51m 00s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2ffef047b2974e269da2961256971cf1 : SUCCESS in 17m 37s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/76b6d34be1a949cd911d62a5e9a1f2f2 : SUCCESS in 31m 06s","accounts_in_message":[],"_revision_number":1},{"id":"0608129f9fd264a24911bbca31a81a3ff7a71f5c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-05 09:38:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c47dd088493da2c2ab49185a1b029b65b1256c1c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-05 13:12:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fafe5664fd772ffb4f55e3a533b4c4a98f374fd3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 13:12:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dd7ce4d4f7c973149c90980c4cbd9eff3fd0153c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:22:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e0023759c5a4b1da0e3955965355eec : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0464a0556250430ab63f5c8114e06d74 : SUCCESS in 5m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/409a1e1e48d94185a328234876bf1259 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c628063ca30d4eeb8e065f3d18a8b063 : SUCCESS in 9m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fca562d7ae89491293bb01a7892d648a : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/48cfb1cf41b1492b94ab154e369fbc26 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3971591202a541d3b0f41f2e1d57cb78 : SUCCESS in 12m 41s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/72e360c5b9df4b5a951e68dbb329724f : SUCCESS in 35m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/88057b2553fe472ba90d46aa37381e0f : SUCCESS in 1h 15m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9820509cc7a642f1af9276d0a771fbdf : SUCCESS in 18m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2db531d64af548c9887d352b079ae4a5 : SUCCESS in 2h 06m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c65febe15f8440c09f6270b9b15691a9 : SUCCESS in 20m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/46dcb98f26c34725995ef570c2689bfe : SUCCESS in 27m 27s","accounts_in_message":[],"_revision_number":1},{"id":"f18da095b39b21419991d89abd41136f6309434a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:22:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8a1d5f26168c65405e9762fcee7396472ef0f1cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:25:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1fb5e5cfbbbe44bbbd30717827ecf14a : SUCCESS in 2m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fea85768209b4ea6950527ed28a0a77e : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1d8b4e2500955ce4385227098ee38c9efbfed038","revisions":{"1d8b4e2500955ce4385227098ee38c9efbfed038":{"kind":"REWORK","_number":1,"created":"2020-04-23 13:21:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/16/722316/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/722316/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/722316/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/722316/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/722316/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/722316/1"}}},"commit":{"parents":[{"commit":"7932314e546d95a1e0a41487096954a3a1310c52","subject":"Merge \"Use upper constraints when installing Tenks\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7932314e546d95a1e0a41487096954a3a1310c52"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2019-05-13 13:02:52.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-23 13:20:59.000000000","tz":60},"subject":"Enable \u0027data\u0027 LVM group only for Docker devicemapper","message":"Enable \u0027data\u0027 LVM group only for Docker devicemapper\n\nThis change is a precursor to switching the default Docker storage\ndriver to overlay2.\n\nPreviously, Kayobe\u0027s default configuration included a \u0027data\u0027 LVM volume\ngroup, with a \u0027docker-volumes\u0027 logical volume mounted at\n/var/lib/docker/volumes. Additionally, if the Docker devicemapper\nstorage driver was used, the data volume group would contain the\ndocker-thinpool and docker-thinpoolmeta volumes.\n\nThis LVM setup was really included for devicemapper, and while in some\ncases it may be useful to have docker volumes as a separate logical\nvolume, this doesn\u0027t really make sense as a default. Often in\nenvironments using other Docker storage drivers, the data volume group\nwould be removed from configuration.\n\nThis change modifies the default LVM configuration to only create a\n\u0027data\u0027 volume group if the Docker storage driver is devicemapper.\nAdditionally, new flags are added to make this choice independent from\nthe storage driver, to support cases where the data volume group is\nrequired without devicemapper.\n\nChange-Id: Ia3c1f6423c32fa2580b57db32512a34ce35d7acc\nStory: 2005667\nTask: 30973\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d8b4e2500955ce4385227098ee38c9efbfed038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1d8b4e2500955ce4385227098ee38c9efbfed038"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
