)]}'
{"id":"openstack%2Fkayobe~723015","triplet_id":"openstack%2Fkayobe~master~Icb57b908be9c83700274c042185e4481d5ca6ea5","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Icb57b908be9c83700274c042185e4481d5ca6ea5","subject":"CI: mark pytest results as a Zuul artifact","status":"MERGED","created":"2020-04-24 19:12:42.000000000","updated":"2020-06-05 13:50:27.000000000","submitted":"2020-06-05 13:19:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723015-1591363171119-1687bc1b","meta_rev_id":"546a1d707d5a6e54bbf0e12f0d8bbdcdc90f7f72","_number":723015,"virtual_id_number":723015,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 13:19:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-05 09:24:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-06-04 06:59:44.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-04 15:16:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-06-05 09:24:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 06:59:44.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-06-04 15:16:12.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-05 09:24:28.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-05 13:19:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8a606766db8fed6b2cb64e7288fc672956146d57","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-24 19:12:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"011fa036b76edcb303f19c6275f83911cf6f07ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 01:37:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af2b008c55c4419dbbc78dfde4b9d177 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd10ffbb97744cfbecd61e7ee52653c : SUCCESS in 8m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/92cefa2330874800a0fcda867cc3619b : SUCCESS in 7m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/458c79682af9453aa1cb5adcc2bdf9fe : SUCCESS in 8m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5e323ef5701149c88c1cdb4d81ef4a5c : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8789d8535fa642e9974b71c67c015a3c : SUCCESS in 11m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8ebcf46c57a24c46b72f6d536af6f15d : SUCCESS in 8m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/901220c2233a42e9b039ea2c8b32f3c3 : SUCCESS in 14m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ab7e758c90b24c05b841980d20d32980 : SUCCESS in 36m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b7929f280fa04ee78b7daba07b4a882e : SUCCESS in 1h 08m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8b9cc758b3194296ad3f66f98f0083f8 : SUCCESS in 16m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fafe1f9b227a41b386deb113d93a1030 : SUCCESS in 1h 55m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e502fef4a90d4fa1ac59b6c000ae0ae7 : SUCCESS in 22m 51s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2542091c115a423f9da6a569e06826bf : SUCCESS in 31m 00s","accounts_in_message":[],"_revision_number":1},{"id":"cea637d5322c6f19536550a9d9257443f37266b9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-25 09:36:21.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"27abe124f1e470b83b674c14615d648eb9dd91d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 11:44:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/149f5e4c19ff49e19b1408cda37095a1 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5cb5a023e704e50a9ff8eb4c94053ca : SUCCESS in 4m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ba365dff136e4df4b9240b3ec272a189 : MERGER_FAILURE in 2s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d9b0e93cfd254a8bb6846ad45d4d4060 : SUCCESS in 4m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a91ef6e8504d470a816c97d19d21ff6b : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f60e27c7b6244b669d44b61e909970bd : SUCCESS in 6m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b1cd1a9e36224f7eb5e48716e8953ad8 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8ab98df415f8429fbc38f2ef31f5bb52 : SUCCESS in 8m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3b33aebb16d74af6808fca327ae61d39 : SUCCESS in 32m 59s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c376d3f55ad64308957243aad2d49134 : SUCCESS in 1h 10m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1464d44bac2d4d319f6f6a3d43a699d3 : SUCCESS in 15m 13s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8d6302765c4b4a309856f39d09466bc5 : SUCCESS in 2h 05m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/50b6d06d5dc643f285c282bfba583c47 : SUCCESS in 17m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ac8f0e69c71c4cf3af1b568b4964ff2c : SUCCESS in 27m 12s","accounts_in_message":[],"_revision_number":2},{"id":"b19df92c9a90a09cd001b9f8b211118fc8144e29","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-25 14:02:03.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"28b1c863e6b233a88f9eb8f0d6d551589f2e25b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 15:53:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/12c58aabcb3c4a23bc39e17b371a4c4e : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a75e6b4b8ec841d28a419e2c6ead2fda : SUCCESS in 5m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ebf3d0ace51d4f4f98daf459dd202116 : SUCCESS in 5m 08s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d530ad36f99144dbaba42ff8424dbc31 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e4028d4bde354eb7a35c1a35b928bd15 : SUCCESS in 4m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5b72a90d75470f845d851f1cb27c3a : SUCCESS in 7m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/89c00311c9e64e078c21e1cf65edc8af : SUCCESS in 5m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/39060ab0be984998883c0b1d6275c128 : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/972e23276e524fcdb5565f078444fd3a : SUCCESS in 26m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8963f8844db8475a8d9050fbd6ba0817 : SUCCESS in 1h 09m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8c661f2ace0c4b78abb047d11d01827d : SUCCESS in 16m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/54ece735df044517b95c253813cd7ea0 : EXCEPTION\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cc7c6d19c5fe480a89e6474c21112a75 : SUCCESS in 16m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5e2c0fcbb2a548f7b32e75833ce54722 : SUCCESS in 25m 28s","accounts_in_message":[],"_revision_number":2},{"id":"148babb38a74f3d1e01abae8acb635b72b92cacc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-25 19:34:50.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b512a508b2b043158182f658e48dda7d : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/096781f257434ae58a1c030cc3b9ecf6 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e36640c3733f416483844d21ea37809d : SUCCESS in 3m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8ad244933fb34b1a94808049a4cbf41b : SUCCESS in 7m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8b8c62bc28b440d9adede48ff7909cec : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578b88060d0c4729938a5fbf2e51e3ec : SUCCESS in 8m 37s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/849534a896a44fb386c1ec2065b32a4e : SUCCESS in 4m 49s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d0a3578fb7a1434da3dd01b2e5cb22dd : SUCCESS in 9m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/54eded1f5b9849df9a63de0845c41d27 : SUCCESS in 28m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9ceef75bcc494b16959280c62753a329 : SUCCESS in 1h 07m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a650e983e34749b993a6fd0ec2b30576 : SUCCESS in 16m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c7d34d990f5f4005bfd7a42f69cd3892 : SUCCESS in 1h 56m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c752b6ae3a3f496b90b6064ab85d0eea : SUCCESS in 17m 56s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/92fd6a265f6247d392612d2c101e0ffc : SUCCESS in 27m 17s","accounts_in_message":[],"_revision_number":2},{"id":"f0533820724d67b4d8f1279aa90545de2a0db0d6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-27 06:23:35.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"714968fd281a99f7f3fa0c3416eb5189696ab4a4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-29 13:49:12.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0ac1d84a7a6465bf987ce43669482ebe19f94986","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 15:49:07.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ea94e4a9618649b28d5fe3b9b3d88a2f : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80491e98f6934888a60bd17fd04019c1 : SUCCESS in 5m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5344d03a87b6460f8ab5f810c5c362c8 : SUCCESS in 5m 07s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3db967b252804f2d8c9cec918902a6ed : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/119398309b3740d78326b2bbcc03009d : SUCCESS in 4m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9bfa41c09a8843e4a5dafab03bf458b7 : SUCCESS in 7m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0d1e34fb9d1941f9b00cbe39f2a8094e : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/73a56ccbc73e4d8895b44605ccc3038a : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4c560f92dd66492b95af4c61d0ccbe0e : SUCCESS in 35m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2d34834e9c0c4188b9e4571d379946aa : SUCCESS in 1h 05m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/20a3070ca5a640898c9b90e316402705 : SUCCESS in 16m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7811de50a54b43068d699fb917089ce0 : SUCCESS in 1h 58m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4d1ae67812994fb69182ca25b1c5573c : SUCCESS in 19m 10s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5b17049eb1274d209f5480850927be42 : SUCCESS in 31m 27s","accounts_in_message":[],"_revision_number":3},{"id":"274bb8b5f377aea9484dd0fbb2a47665cd170dd8","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-04 06:59:44.000000000","message":"Patch Set 3: Code-Review+1\n\nNice, wasn\u0027t aware of that role.","accounts_in_message":[],"_revision_number":3},{"id":"9959d3d74341e3156fded4c293ee9cf87299e269","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-04 15:16:12.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"45b71b86586e80b31e966f11702542db9c427526","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-05 09:24:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3888ab363ac212dd9be56cae34851d35b587e0e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 09:24:41.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e5fb8f53fc60e9bb04a1ac013537b2fb131ab1b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 13:19:30.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6053fce29aa741cb97f0b4d02cf919b6 : SUCCESS in 6m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2a8fc5cef85740359567cdc6758d2bda : SUCCESS in 6m 16s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f88dd99aee60479089c2018d92a3ce58 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/481d66d3a1504f048ea565d10944759b : SUCCESS in 8m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a394ca6c3af1417eb2520830778a3964 : SUCCESS in 5m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5b81ad0b5f5c4389b4485187bc9f45f0 : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a931c6d87817406a84277779a1c13d2d : SUCCESS in 36m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3bd15ec9915946d6ab7c414a2ca4f5da : SUCCESS in 1h 14m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/442a8c52575a45fcbb9062073cd7c09f : SUCCESS in 20m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e5959027126740a7ab22de523b5f8abe : SUCCESS in 1h 28m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/613f10e48da441a4ab55400d182608ff : SUCCESS in 17m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a151c08bec934d0c9ba64b2661d01c2c : SUCCESS in 29m 44s","accounts_in_message":[],"_revision_number":3},{"id":"f4ae5ea075e94ac9f4f466abae0258f8c2aa653b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 13:19:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fae0c541a22ad9818aada0d3ce9fd34f2b81ac66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 13:20:53.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d563cd9be84e4db0a1f921e5573cd843 : SUCCESS in 1m 03s","accounts_in_message":[],"_revision_number":3},{"id":"546a1d707d5a6e54bbf0e12f0d8bbdcdc90f7f72","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-05 13:50:27.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 7bb215eac4607be530283e657ed57a86bcc94950","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b49edbd2752bf7c4a6c81fa18a454e2e3f4bc1bf","revisions":{"4aabd365545465cf013338a72c6c00530574c8ff":{"kind":"REWORK","_number":1,"created":"2020-04-24 19:12:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/723015/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/723015/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/723015/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/723015/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/723015/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/723015/1"}}},"commit":{"parents":[{"commit":"558276a8a6af4a7375bf58afd7ab033e11a93296","subject":"Merge \"CI: Add overcloud host configure job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/558276a8a6af4a7375bf58afd7ab033e11a93296"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 19:12:31.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 19:12:31.000000000","tz":60},"subject":"CI: mark pytest results as a Zuul artifact","message":"CI: mark pytest results as a Zuul artifact\n\nChange-Id: Icb57b908be9c83700274c042185e4481d5ca6ea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4aabd365545465cf013338a72c6c00530574c8ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4aabd365545465cf013338a72c6c00530574c8ff"}]},"branch":"refs/heads/master"},"140f0460ba3f26a3a6a0902fe7264d5bea7242a2":{"kind":"REWORK","_number":2,"created":"2020-04-25 09:36:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/723015/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/723015/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/723015/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/723015/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/723015/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/723015/2"}}},"commit":{"parents":[{"commit":"558276a8a6af4a7375bf58afd7ab033e11a93296","subject":"Merge \"CI: Add overcloud host configure job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/558276a8a6af4a7375bf58afd7ab033e11a93296"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 19:12:31.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-25 09:36:18.000000000","tz":0},"subject":"CI: mark pytest results as a Zuul artifact","message":"CI: mark pytest results as a Zuul artifact\n\nChange-Id: Icb57b908be9c83700274c042185e4481d5ca6ea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/140f0460ba3f26a3a6a0902fe7264d5bea7242a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/140f0460ba3f26a3a6a0902fe7264d5bea7242a2"}]},"branch":"refs/heads/master"},"b49edbd2752bf7c4a6c81fa18a454e2e3f4bc1bf":{"kind":"REWORK","_number":3,"created":"2020-05-29 13:49:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/15/723015/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/15/723015/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/723015/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/723015/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/15/723015/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/15/723015/3"}}},"commit":{"parents":[{"commit":"f2c21143709d9fa752c2c694fb4dd64c0ff5010c","subject":"Merge \"CI: Encrypt passwords.yml in overcloud host configure job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2c21143709d9fa752c2c694fb4dd64c0ff5010c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 19:12:31.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 13:49:03.000000000","tz":60},"subject":"CI: mark pytest results as a Zuul artifact","message":"CI: mark pytest results as a Zuul artifact\n\nChange-Id: Icb57b908be9c83700274c042185e4481d5ca6ea5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b49edbd2752bf7c4a6c81fa18a454e2e3f4bc1bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b49edbd2752bf7c4a6c81fa18a454e2e3f4bc1bf"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
