)]}'
{"id":"openstack%2Fkayobe~723262","triplet_id":"openstack%2Fkayobe~master~Ie42900d89afff40b8a12364b38de9854ed817275","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ie42900d89afff40b8a12364b38de9854ed817275","subject":"CentOS 8: add docs and release note","status":"MERGED","created":"2020-04-26 19:59:40.000000000","updated":"2020-05-28 15:59:35.000000000","submitted":"2020-05-28 15:59:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723262-1590681575590-41f9ad85","meta_rev_id":"beb1294ca9c434e233b666e6f85b5d2c62e1a4ea","_number":723262,"virtual_id_number":723262,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 15:59:35.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-05-28 09:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-28 09:26:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-05-28 09:26:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-05-28 09:26:42.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 07:41:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-29 15:37:22.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-05-28 15:59:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a3d45803d19e27c96e4b0d72e6302b901215e499","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-26 19:59:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"400dcebe1c0bd6135b3da72516beaeab5da3eb21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-26 20:07:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/953976d6b7eb4c26b1f611dcd621953c : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e21bb5ca3c2646f9920708bceffcc2be : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b63177ef541e4490a9bbdb630ed2a2c0 : SUCCESS in 3m 43s","accounts_in_message":[],"_revision_number":1},{"id":"e4a39255ea92b5b1d8b000cfcd735dfdbf030e75","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 14:05:21.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"dee22b611961a7af43cc33e002224231fbdbde7c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 14:06:14.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"a3ac797ee5a41ccebb996d050b72fdb9c81bbca0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 14:24:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a90cc00d3d7b4516affc9db3481ff173 : SUCCESS in 7m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72507e83c0c4459199c38a0617b44372 : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/24b12e4b2927407c8a3df8e91b4b0071 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":2},{"id":"dd743c920824e45f9d0a77285eee5ba977a3ca13","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-28 07:41:51.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e2a4b3fa87efb5d35d1ef64a673cc6558af0180c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-29 15:37:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7813c77841c24455b4f67694c92fd550b896e57f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 10:03:07.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"0d3232726836d3efa407b4e6889eed4a1bbd3d55","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-30 10:03:14.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"250cff70ac0ba4989a157535310bf41e7f181a72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 10:10:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/398b79371beb4f58b69576c62ea5987a : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/521b9c8824d6442eb53f3434670f2ade : SUCCESS in 6m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e6ca1cf05118491c9a843775b22e6189 : SUCCESS in 3m 44s","accounts_in_message":[],"_revision_number":3},{"id":"9aa9b1ab2f9ff31c36dbfd2ac8e62f9eea0dbd47","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 10:43:49.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"03cb23c6a8f52c45bb4986b3acdc28ff2d02a8c0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 10:43:56.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"eececc23e0ec630783d0ed9dca8c269dc5487630","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 10:56:27.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/547730a128994c4cb9247f7c7b4efa76 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6a15159cf3848a0b66613afa210694c : SUCCESS in 11m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b755c153aa1541df826e84e3ba655d43 : SUCCESS in 3m 36s","accounts_in_message":[],"_revision_number":4},{"id":"d77bf5600e64c099067a495602549edcd4fe5cdb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 08:12:51.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"35dd2fecaa1622033d25bdc2fb25b850bfe1f088","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 08:18:47.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"06d3a5a42a85d3db5629886974d431cf2b23ab53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 08:29:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93d116921c14481cb5a52b74640be70a : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783b77f372ec4781a2dd1aaa74fa7207 : SUCCESS in 8m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dfca83ee0d14c078deecac7367467a0 : SUCCESS in 7m 14s","accounts_in_message":[],"_revision_number":5},{"id":"9da28a05bb2cd07f4de3ee5011a4942b161d9640","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 15:39:21.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"c897123f521bee5d00aa0a6b19d518e957022f09","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-19 15:39:31.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"9d92f5e50c347d6d176e0b6d041e3c76f6541813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-19 15:59:48.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa60613e42604c5ba900788076cb31de : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/879bcf3aa22a4e59ab2ce27e6b3b38b8 : SUCCESS in 13m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cb3ce6c7ff874a52af31bc740a9694b6 : SUCCESS in 7m 13s","accounts_in_message":[],"_revision_number":6},{"id":"4525775161217401a06cd36f8798663da33e13ea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 09:26:42.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"7682c65028d334c339fe83e952c1ff7b4ac34ea5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 09:26:56.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"1b588281cb42a479d8317af1ce0475a9521d080a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 09:42:23.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d776d64104c849e8ab29899978031697 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac53f461dfee48fc9f33d721a696d387 : SUCCESS in 8m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29af62c5bce1465eae403eceefa4fead : SUCCESS in 6m 07s","accounts_in_message":[],"_revision_number":7},{"id":"42a369826c765c160286c572fc75c4bd12ee3310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:07:14.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3eca7053efa240d6dd05b1440845b0a2abe86be7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:59:35.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5688466bd8834eda8f89ac64e5bec500 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79a39dbe9530471b8f010925d5a0f143 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c47eba8c42f448394f8b61e09d24e6f : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":7},{"id":"19c88bac656f1d36ca30a5a9e4508fee385a1e61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:59:35.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"922175ebd223a2a747dd70f78dfde54e81f3a4e9","revisions":{"cf3d759f164f612ee6445f9e77098f1a214a1eee":{"kind":"REWORK","_number":1,"created":"2020-04-26 19:59:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/723262/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/723262/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/723262/1"}}},"commit":{"parents":[{"commit":"78c8a70c5d61dcd0924298050329f58b944f9093","subject":"Remove support for CentOS 7 and Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/78c8a70c5d61dcd0924298050329f58b944f9093"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf3d759f164f612ee6445f9e77098f1a214a1eee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cf3d759f164f612ee6445f9e77098f1a214a1eee"}]},"branch":"refs/heads/master"},"a9c53348d1d8c221700d62515ca2df4af2e6e068":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-27 14:05:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/723262/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/723262/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/723262/2"}}},"commit":{"parents":[{"commit":"658731fa1aa96c4f9ef8f946cd9113bd0a25d948","subject":"Remove support for CentOS 7 and Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/658731fa1aa96c4f9ef8f946cd9113bd0a25d948"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-27 14:05:10.000000000","tz":60},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9c53348d1d8c221700d62515ca2df4af2e6e068"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a9c53348d1d8c221700d62515ca2df4af2e6e068"}]},"branch":"refs/heads/master"},"c20ddc75b278416b576dfca191d904d8fec46fbd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-04-30 10:03:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/723262/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/723262/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/723262/3"}}},"commit":{"parents":[{"commit":"4c785437d2972f8543d3f5b2f60b2a435113f51b","subject":"Remove support for CentOS 7 and Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c785437d2972f8543d3f5b2f60b2a435113f51b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-30 10:03:07.000000000","tz":0},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c20ddc75b278416b576dfca191d904d8fec46fbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c20ddc75b278416b576dfca191d904d8fec46fbd"}]},"branch":"refs/heads/master"},"291cd65dfbcecefa51b65793c88f2a4d938f0d48":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2020-04-30 10:43:49.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/62/723262/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/723262/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/723262/4"}}},"commit":{"parents":[{"commit":"9e7e6e0a7f6a1dc1c1d5f68cd658a51c56cc5d10","subject":"Remove support for CentOS 7 and Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e7e6e0a7f6a1dc1c1d5f68cd658a51c56cc5d10"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-30 10:43:49.000000000","tz":0},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/291cd65dfbcecefa51b65793c88f2a4d938f0d48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/291cd65dfbcecefa51b65793c88f2a4d938f0d48"}]},"branch":"refs/heads/master"},"5942b3bd5d5cff1fa177aeca07e682e49c751861":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-05-19 08:12:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/723262/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/723262/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/723262/5"}}},"commit":{"parents":[{"commit":"589f317eb0057dbd0fd0fb03904cc019c91df6fe","subject":"Remove support for CentOS 7 and Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/589f317eb0057dbd0fd0fb03904cc019c91df6fe"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-19 08:12:39.000000000","tz":60},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5942b3bd5d5cff1fa177aeca07e682e49c751861"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5942b3bd5d5cff1fa177aeca07e682e49c751861"}]},"branch":"refs/heads/master"},"0caa69abe495e99d786143067aadfb3c8eb46bd3":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-05-19 15:39:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/723262/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/723262/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/723262/6"}}},"commit":{"parents":[{"commit":"38f1425d35bd2f6126123e0865542f20e684f0f3","subject":"Remove support for CentOS 7 and Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38f1425d35bd2f6126123e0865542f20e684f0f3"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-19 15:39:21.000000000","tz":0},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0caa69abe495e99d786143067aadfb3c8eb46bd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0caa69abe495e99d786143067aadfb3c8eb46bd3"}]},"branch":"refs/heads/master"},"922175ebd223a2a747dd70f78dfde54e81f3a4e9":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-05-28 09:26:42.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/62/723262/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/62/723262/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/62/723262/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/62/723262/7"}}},"commit":{"parents":[{"commit":"b9d76f6ef5e9ada68df47a417a50708a262005b1","subject":"Remove support for CentOS 7 and Python 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b9d76f6ef5e9ada68df47a417a50708a262005b1"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-28 09:26:31.000000000","tz":60},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/922175ebd223a2a747dd70f78dfde54e81f3a4e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/922175ebd223a2a747dd70f78dfde54e81f3a4e9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
