)]}'
{"id":"openstack%2Fkayobe~723341","triplet_id":"openstack%2Fkayobe~master~I2b6d62104c9c127cb8f6b4f4930dd695cd00da17","project":"openstack/kayobe","branch":"master","topic":"monasca-defaults","hashtags":[],"change_id":"I2b6d62104c9c127cb8f6b4f4930dd695cd00da17","subject":"Define default variables for Monasca config","status":"MERGED","created":"2020-04-27 08:43:53.000000000","updated":"2020-07-23 09:27:03.000000000","submitted":"2020-04-29 17:33:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723341-1588181628899-276f548f","meta_rev_id":"8686cdb6759ceae2cfef827e57de5f5eb7e229e4","_number":723341,"virtual_id_number":723341,"owner":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-29 17:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-04-29 15:32:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":2,"date":"2020-04-28 13:56:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-04-29 15:32:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 13:56:59.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-04-29 15:32:50.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-04-29 17:33:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2977bb23c639066a542faa917cde31ada8e08efc","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-04-27 08:43:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61bbe7e4514c144468d880df3c38d3b6c3bfd12f","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-04-27 09:02:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"601a0f5a09af6c12e095ccf3687b5bd47e7b9729","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:00:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ddb15ec0177430aba9e16b35ac575f8 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49adf79fd04942c580e69b84fd6498ac : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b581a4d3ef0c4e81b69becb363260860 : SUCCESS in 5m 34s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/401387a26ca7499b89e5984c8c2a43f4 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f6ea1fdced2f44bdb2240a0646623c7e : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b735d89bc62a4119ba00255317ba34f9 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/712b662e655e43f78678d772a9453089 : SUCCESS in 4m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b49f6bc073ea499a920ebc72e3caac61 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/693093f90b8a4ae1ab4c968c0cde980f : SUCCESS in 10m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2699c89bafcd45569254dd61a43e6619 : SUCCESS in 33m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a807ce953f57446895f5dd54c2c24bc2 : SUCCESS in 1h 05m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8b2c69ce0c974512a2ad53275d4b7c2b : SUCCESS in 16m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/875c1e99ea284829a83ec1da15af9978 : SUCCESS in 1h 53m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cf41013b9e254d8e8775ba3c0322231c : SUCCESS in 17m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7331c9ec5462440e81053dcfa1d23c41 : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":2},{"id":"8837d19dc71122d2ef0e844a79472ba97ef36cc5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 11:09:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"196a2206fe6dde5d63835b6a605c3877e0110b37","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-04-27 14:49:18.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ac2935710a6b079a66615bb0e7c5068170b77db7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 15:06:25.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"ed4d30577fb0233b9857bb1c83b9d2d72972b189","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-04-27 19:53:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5f832755cc43c4710b0cf10ed7c318525765d796","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 21:50:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0466fe3a7f914792a0f224105aa3983f : SUCCESS in 5m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/062db134bbbf4e579467536452ba4443 : SUCCESS in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6cd13354d4aa4a86b052ea3d5bb91f40 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/98961956745748e092fe42376df69371 : SUCCESS in 6m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2414b97aa9e84359a715d89b07146260 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bfd6356eb6a404a9801eb87aaf825bf : SUCCESS in 7m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f8d24bf5764444dea28618119a630dc6 : SUCCESS in 5m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b6d294ac0ecb46659f18c2cd7751cb2e : SUCCESS in 6m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a841817f28e54ef88ec7b37fbe36f922 : SUCCESS in 10m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e77d50b4eaa94361be6f0ac4e0c1e552 : SUCCESS in 34m 45s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/da54e6d00bdb4c31b981a15865f03874 : SUCCESS in 1h 09m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6e1e56eaf4e34d7397d729b8a4e0750f : SUCCESS in 16m 45s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/17bf298ec79f4cbca2021c64e041812c : SUCCESS in 1h 50m 59s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/efce95a4284c474f84db7b701bf856cf : SUCCESS in 19m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d7ba1ec95b234d32afcb4968c4b20512 : SUCCESS in 30m 01s","accounts_in_message":[],"_revision_number":3},{"id":"744fa6ff8f4fa54371308b1ee4d954b6c24fc729","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-28 11:31:52.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"666752da292ddd68207940ec4880a1dac59b814b","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-04-28 13:50:47.000000000","message":"Uploaded patch set 4: Commit message was updated.","accounts_in_message":[],"_revision_number":4},{"id":"fea482fac6e08247805d6fac6e881145ea947edf","author":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"date":"2020-04-28 13:53:53.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"9a89a48298a4f7ce508ac71fbce9f9054f81b4c4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-28 13:56:59.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks Isaac.","accounts_in_message":[],"_revision_number":5},{"id":"c62fa488d1e18294e6a4dd0d78d47a5adc81bfb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 16:03:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8f75d131f784d3ebdd1d735a1f46116 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31fd11ead9df4b77b6277e3621225320 : SUCCESS in 8m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdd2f34b3f2d4997b69969b5a0668747 : SUCCESS in 4m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/5848198c39344005ad8eb8cc12fdce2a : SUCCESS in 11m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48952b03d804427bb8e4e58c3cfb1eee : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/003dd134406a41eebf458c06e3fb3bb2 : SUCCESS in 6m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba41d6ff43894eb9bc290d95fc7e3294 : SUCCESS in 4m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7fd638846f6d4e6da5a4577a2fe162ec : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ef919145ab94839ac933520ed56c7a9 : SUCCESS in 10m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/46118a5b5311421d980cc602bcd4d752 : SUCCESS in 36m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8211d84e1ca240a498fcf155d82d5416 : SUCCESS in 1h 09m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b283dc19a62d434a9a7cd23aaf42002f : SUCCESS in 15m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e481e02634ca40759584e9f38093137e : SUCCESS in 2h 04m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/edfa460a5978423791a37b31d6c027f3 : SUCCESS in 19m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/35c95a0b8a0c437cacfad20eca4c6ab3 : SUCCESS in 31m 07s","accounts_in_message":[],"_revision_number":5},{"id":"9d3a867345adcf9a73dba5855139f83842eec5a4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-04-29 15:32:50.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"401a43f9d55e4b75b6b39b9d82352d743bc23c89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 15:33:26.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a6b1bf32fc2f4cb2e92ed71a987b4c2fd1152d07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 17:33:48.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d198922bb48a4e5db68ed22a8b6fd711 : SUCCESS in 6m 03s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9fb0a3ffb1724df9b396bab6c63a7c73 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0614ecf8a4b4e8fa5c79775dd3b964f : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f148df0f68084fe8ac63b12c01d88801 : SUCCESS in 7m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a50214d0a42a4a3c9e0320758d3bc004 : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2d077b428f954a96a14ed34e37367ef5 : SUCCESS in 5m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1bb393f166df407795286e0640463a8a : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be6d7ab4bebd44ca9bac2e8f1d730d90 : SUCCESS in 30m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5a9178f932154a279ff8823b1be26199 : SUCCESS in 1h 13m 44s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a6b56093117140eb931f9c04a98b5c57 : SUCCESS in 17m 49s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/669521e683584bff923d0df89b0eb4ef : SUCCESS in 1h 56m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a0fa0e199ca642c18f6b5cade9ee94bf : SUCCESS in 17m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d8d4366c00794ed38285bd77ef47ca66 : SUCCESS in 26m 27s","accounts_in_message":[],"_revision_number":5},{"id":"ab5c29672b95ecbc0deff60ee7c464dbd951da03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 17:33:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"4e8acfea0d80cd4012dca40240c0b4373bf7dfe2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 17:37:13.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7157bb2c2a6d479a9b7ce6e21912235a : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1cae3c51fdcd428a949b5384c0c77813 : SUCCESS in 1m 16s","accounts_in_message":[],"_revision_number":5},{"id":"72b4020aad47de538d95501258a59d6069358abf","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-22 17:12:59.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit 79b43f951b4d4498b7506fef37b81b493ef9729c","accounts_in_message":[],"_revision_number":5},{"id":"867405aa829d32240cc67755f827c3f67d13dba8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-22 17:14:58.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit b47a7616eff53a0e66b144aab2913990e7cbf0fe","accounts_in_message":[],"_revision_number":5},{"id":"f9df1420369c3b86206fd09386c332a96fdacf35","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-22 17:15:29.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit f7e4d37b45bba872169df1f02991cfbe114307ca","accounts_in_message":[],"_revision_number":5},{"id":"0a5f67f0f123880528bef3ad839abc72822d4ebb","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-07-22 17:16:56.000000000","message":"Patch Set 5:\n\nI\u0027ve cherry-picked this one as I cannot override the hardcoded port in the grafana-url, so I consider this a bug.","accounts_in_message":[],"_revision_number":5},{"id":"5edd150eba43222b80ff21df6eb89e5769c5ceea","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-23 09:27:03.000000000","message":"Patch Set 5:\n\n\u003e I\u0027ve cherry-picked this one as I cannot override the hardcoded port\n \u003e in the grafana-url, so I consider this a bug.\n\nShould be possible to set grafana_conf_grafana_url.","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"39308a45914cc452362b6f4442785ddc7e0c26c7","revisions":{"bda67d740c716a2bd7bebb012f53293ad28c8160":{"kind":"REWORK","_number":1,"created":"2020-04-27 08:43:53.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/41/723341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/723341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/723341/1"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-22 15:29:01.000000000","tz":0},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-27 08:23:59.000000000","tz":60},"subject":"Define default variables for Monasca config","message":"Define default variables for Monasca config\n\nUse default Monasca parameters to flexibly configure\nGrafana post deploy.\n\nChange-Id: I2b6d62104c9c127cb8f6b4f4930dd695cd00da17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bda67d740c716a2bd7bebb012f53293ad28c8160"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bda67d740c716a2bd7bebb012f53293ad28c8160"}]},"branch":"refs/heads/master"},"44ffae439150da7bab50e27d77760f276fe19a2e":{"kind":"REWORK","_number":2,"created":"2020-04-27 09:02:02.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/41/723341/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/723341/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/723341/2"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-22 15:29:01.000000000","tz":0},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-27 09:01:40.000000000","tz":60},"subject":"Define default variables for Monasca config","message":"Define default variables for Monasca config\n\nUse default Monasca parameters to flexibly configure\nGrafana post deploy.\n\nChange-Id: I2b6d62104c9c127cb8f6b4f4930dd695cd00da17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44ffae439150da7bab50e27d77760f276fe19a2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/44ffae439150da7bab50e27d77760f276fe19a2e"}]},"branch":"refs/heads/master"},"9e31f9a8b62b735db939253429089865554c6821":{"kind":"REWORK","_number":3,"created":"2020-04-27 19:53:04.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/41/723341/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/723341/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/723341/3"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-22 15:29:01.000000000","tz":0},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-27 19:52:48.000000000","tz":60},"subject":"Define default variables for Monasca config","message":"Define default variables for Monasca config\n\nUse default Monasca parameters to flexibly configure\nGrafana post deploy.\n\nChange-Id: I2b6d62104c9c127cb8f6b4f4930dd695cd00da17\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e31f9a8b62b735db939253429089865554c6821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9e31f9a8b62b735db939253429089865554c6821"}]},"branch":"refs/heads/master"},"ceb72b1192d94ec46328a83e7a11fd7c21aea388":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2020-04-28 13:50:47.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/41/723341/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/723341/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/723341/4"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-22 15:29:01.000000000","tz":0},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-28 13:35:43.000000000","tz":60},"subject":"Define default variables for Monasca config","message":"Define default variables for Monasca config\n\nUse default Monasca parameters to flexibly configure\nGrafana post deploy.\n\nChange-Id: I2b6d62104c9c127cb8f6b4f4930dd695cd00da17\nStory: 2007597\nTask: 39587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ceb72b1192d94ec46328a83e7a11fd7c21aea388"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ceb72b1192d94ec46328a83e7a11fd7c21aea388"}]},"branch":"refs/heads/master"},"39308a45914cc452362b6f4442785ddc7e0c26c7":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-04-28 13:53:53.000000000","uploader":{"_account_id":30438,"name":"Isaac Prior","email":"isaac@stackhpc.com","username":"Wasaac"},"ref":"refs/changes/41/723341/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/723341/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/723341/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/723341/5"}}},"commit":{"parents":[{"commit":"e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7","subject":"Update inventory templates for Ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e0932bd788762e38d6c55c34eeae2fd6c8e5d0c7"}]}],"author":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-22 15:29:01.000000000","tz":0},"committer":{"name":"Isaac Prior","email":"isaac@stackhpc.com","date":"2020-04-28 13:51:26.000000000","tz":60},"subject":"Define default variables for Monasca config","message":"Define default variables for Monasca config\n\nSets \u0027monasca_install_type: source\u0027 to remove need\nfor kolla-ansible var boilerplate.\n\nAlso use default Monasca parameters to configure\nGrafana post deploy.\n\nChange-Id: I2b6d62104c9c127cb8f6b4f4930dd695cd00da17\nStory: 2007597\nTask: 39587\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39308a45914cc452362b6f4442785ddc7e0c26c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/39308a45914cc452362b6f4442785ddc7e0c26c7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
