)]}'
{"id":"openstack%2Fkayobe~723352","triplet_id":"openstack%2Fkayobe~stable%2Frocky~I77c94edf654565a12ce8be681e3c9b16caa55c86","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"I77c94edf654565a12ce8be681e3c9b16caa55c86","subject":"Fix loading of Monasca Grafana dashboards","status":"MERGED","created":"2020-04-27 09:06:46.000000000","updated":"2020-04-30 17:14:48.000000000","submitted":"2020-04-30 17:14:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723352-1588266888349-71cafbf4","meta_rev_id":"1d57d5d7d84b9fdbd05a74d645173f62310943c6","_number":723352,"virtual_id_number":723352,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 17:14:48.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-28 20:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-04-30 15:20:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-28 20:30:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-30 15:20:54.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-04-30 17:14:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9008240d0beeb6003ae4167caecfd78451472ce8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 09:06:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1136a2cbed75daf4e111688007fbf5ab8c35b79b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 10:57:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c8a1ba1af33c447cbc656e2c73187530 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85b7fd7568f4e688f0882786e6369b1 : SUCCESS in 6m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a1220d6c45c442d8e620b7c21365585 : SUCCESS in 7m 29s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3b8b3bc4f9bd42c78dfc5ca34ed05f87 : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bab94f5d2824896aebb61bc1c62846b : SUCCESS in 5m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4bda93f62c924b2799b14da209c9a767 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd4af0d71f824faeaae5ad17817b2c1b : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f594284863e64aa19d05df8c65a77c6c : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2de188b3fccd4a7583a2b09b09966d36 : SUCCESS in 10m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bff56433506b47248e15870c707c6895 : SUCCESS in 31m 56s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/0546b73eaf99489097cd940e496c7bb1 : SUCCESS in 1h 01m 22s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/646d9c51cecb4672bbad51abff8498c2 : SUCCESS in 1h 49m 30s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/87cca6500804473e970b02273ea422d3 : SUCCESS in 19m 35s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/fc1475d0ffee46ddab1bc1f3799c0e6b : SUCCESS in 33m 50s","accounts_in_message":[],"_revision_number":1},{"id":"3049aa5265a0c41e4879cb5fed70239883653f5b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-28 20:30:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2c889896b42ce4cabab7dfb781ae27cbcd6d6b2d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-04-30 15:20:54.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"42668d4bcb98b2351b9bfd2f4d43fc17a9612cb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 15:21:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c19e2bdbdbf19cd42336a798453939a37a332c8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 17:14:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"1d57d5d7d84b9fdbd05a74d645173f62310943c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 17:14:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/b3c07672d08447ab8ca96643eb0cd692 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f55829b62d449fa8877be9d7d870827 : SUCCESS in 7m 50s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/353c52b82bd14f4cb8c49633e1fd4053 : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71134f8ef0f2415596719d930966db86 : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e88228afe97549d08f6adfaa82ff666c : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8fadcf9a6892435eb908513029f2f2b2 : SUCCESS in 5m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/82d98ab71c4745c3a087a7437459a113 : SUCCESS in 6m 37s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3687d6a3465a489c92930fb5f60e692d : SUCCESS in 9m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7006a92d2ec4e2496794190ced2ed06 : SUCCESS in 27m 38s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/319bc8d6523141ba83590977fe038c15 : SUCCESS in 1h 12m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/0b3915bd34494b46bf714b2f60b6b46c : SUCCESS in 1h 41m 59s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/ea1a2c151a224e29b4d3d23de642c0e9 : SUCCESS in 20m 45s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/9e4ca45ffdc24b04bfc074fff6ff1885 : SUCCESS in 38m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"abeb84f377dca9a0f71ba73e05b30d0b3b81a9f0","revisions":{"abeb84f377dca9a0f71ba73e05b30d0b3b81a9f0":{"kind":"REWORK","_number":1,"created":"2020-04-27 09:06:46.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/52/723352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/52/723352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/723352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/723352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/52/723352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/52/723352/1"}}},"commit":{"parents":[{"commit":"d9a6a60ae178a06fbed87c537cba40f0534a97ff","subject":"Fix seed VM provisioning on a remote seed hypervisor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d9a6a60ae178a06fbed87c537cba40f0534a97ff"}]}],"author":{"name":"Doug Szumski","email":"doug@stackhpc.com","date":"2020-03-28 17:08:38.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-27 09:06:36.000000000","tz":60},"subject":"Fix loading of Monasca Grafana dashboards","message":"Fix loading of Monasca Grafana dashboards\n\nThis is a minimal fix to support loading dashboards into the Monasca\nGrafana fork. It firstly aligns the default Monasca Grafana control\nplane organisation and Monasca Grafana local admin username with Kolla\nAnsible to make the feature easier to use. Secondly, it extracts the\nassociated OpenStack project name from this variable by stripping off\nthe OpenStack domain.\n\nLonger term we may wish to move the dashboard loading functionality into\nKolla Ansible, now that it supports deploying Monasca.\n\nAffects Rocky onwards.\n\nChange-Id: I77c94edf654565a12ce8be681e3c9b16caa55c86\nStory: 2007477\nTask: 39186\n(cherry picked from commit cdc1d5ada2283a742ee5946b54ac2f87efa5e431)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abeb84f377dca9a0f71ba73e05b30d0b3b81a9f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/abeb84f377dca9a0f71ba73e05b30d0b3b81a9f0"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
