)]}'
{"id":"openstack%2Fkayobe~723355","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Ib477de5f3af2e4c182d0c2999c274dbb5553531c","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"Ib477de5f3af2e4c182d0c2999c274dbb5553531c","subject":"Use upper constraints when installing Tenks","status":"MERGED","created":"2020-04-27 09:11:00.000000000","updated":"2020-04-27 18:05:51.000000000","submitted":"2020-04-27 18:04:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723355-1588010643942-81340386","meta_rev_id":"fe424a6346384982349cbd1b7a143137966c42d9","_number":723355,"virtual_id_number":723355,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-27 18:04:03.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-27 11:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-27 09:44:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-27 11:04:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-27 09:44:00.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-27 18:04:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"185fe5a75e58e380aca365f0730ee57826faaf48","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 09:11:00.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"67d35659a228ed7f3ba9cc5e16993554948fe221","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-27 09:44:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"818d5ce025c3ecd3c53186358a6772e1c8303f67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 10:51:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6d01464340145d5be43a0c418466f32 : SUCCESS in 6m 34s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f3b2510f1d5c469a9c567d6678e3df14 : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d2a1e2aa60714ca9aadb3358214a7bf8 : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6db7407fef4d22ad0830b194536b91 : SUCCESS in 6m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e69955e13bc14a03843aacbfc3866f88 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/066b724b52d5472abbead9f197f59463 : SUCCESS in 5m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c929aa9185314d34ba84700ab78d41d1 : SUCCESS in 5m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fdf70158a3ae4a128e043a9eafec49b0 : SUCCESS in 8m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1905819b2945447798f372da2b7f26ea : SUCCESS in 35m 14s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/b3b2ba2252784da38d5faff5d020f3fd : SUCCESS in 1h 09m 47s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/b2653c1c1cc546799acd0988190f307d : SUCCESS in 1h 40m 11s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/3895b4cabffd4912a71db6a9901f81eb : SUCCESS in 23m 23s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a36223ddc68b4182a5e4a1cd28befd41 : SUCCESS in 35m 38s","accounts_in_message":[],"_revision_number":1},{"id":"2a283776ea0705b64e7f966302725ed3939f92b5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 11:04:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"404e8a2af60817d50f70f2f223271307f53e53cb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 11:04:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ab394c4142fb7fcb447b6efff8083cb54e8a5f89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 12:46:48.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/460f63b409a8424da11b3b12319dff66 : SUCCESS in 4m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/66583f8e2da048a49be0d42723663ef0 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5f764cb4e5f46debaed6b1a08e38133 : SUCCESS in 4m 40s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/0989072351ea40779e40c9c528b9807e : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84766b66527f4849bc936fdf6a0640a7 : SUCCESS in 4m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7a059e4dc839463db6733e8b10001a89 : SUCCESS in 4m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/374ca17efe864a91bc2514734b5f5753 : SUCCESS in 8m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f76236ad9139479fb042ed41ed496dd9 : SUCCESS in 28m 37s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9ff067338f6e4c5b96b348a1c78b8f7e : SUCCESS in 1h 01m 11s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3ba81ed3e6c74478a5011722acba0df4 : SUCCESS in 1h 41m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/95d9e80980524cfdbc0d7858afb9b945 : SUCCESS in 20m 48s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4564d1910bb9407cbca35faed3ceb29c : FAILURE in 19m 12s","accounts_in_message":[],"_revision_number":1},{"id":"d7594c76b4e69f87149fd13467e5c3545cf3db67","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 13:30:18.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"06b86ede9fafedd834f0542e9c549b32b5f66a5a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:57:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e712776acf314a3eb845f8e1d67c7e5f : SUCCESS in 7m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f758b4d8f34047afaee1046f989ea5bf : SUCCESS in 7m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b37b96b6fa849fc936f07f34623d77f : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1f12b17d09c404f94cfbbb4f1dc91e6 : SUCCESS in 6m 12s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16d6a25f9a3f4cf2933c72c8e72897e5 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/299eb63fb0d24ec8b826be075d686afc : SUCCESS in 6m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cfe48f406d514c7bb2eee953bcfaaafa : SUCCESS in 6m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/50c222618a54433d9d4c3c5354106ad5 : SUCCESS in 10m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/acc1a09afd4b486a9219f93ede313de0 : SUCCESS in 35m 31s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/eba7e2c4e2f140a6a9a574ab9934d7b2 : SUCCESS in 1h 08m 50s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/bf40f40ccce34938a03c134b86bfc99b : SUCCESS in 2h 05m 00s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/dfc4a811f0254e5892d1cf3a64c66671 : SUCCESS in 23m 44s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/a70662d218be422e89bcbcca8ce659b6 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c799fdfd1c6c6ef92bb0f3cd038b723a8f793cec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:57:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c715e963b811765b1f0247b6b5e62a91c2769d34","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:04:03.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2fa90b194363cb086b389e0146e55daa412dc4dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:04:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/74e5274f513a4820a585aea77bab7553 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d489efd2020d41debd405654e748e0e3 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac2e3bd84b9e466db230b8f236cbcc20 : SUCCESS in 4m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e50bb237dabd4cb48343c98ac35775c0 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecacd24381e64c6d9c2cc1d52e88da48 : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8b944887748a4b048b6d25421563f6ac : SUCCESS in 5m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6bb53b80e2ae486eaf121719c8f75d04 : SUCCESS in 9m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/91a97a190a844ac89887049331bd5598 : SUCCESS in 36m 14s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/360285dc2d034e679156e3da5298c3b5 : SUCCESS in 1h 14m 27s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7982f151d49a41c3b9606632a9ed3cee : SUCCESS in 1h 58m 27s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0a8153f1e4d34763b6b567f8f24da6b7 : SUCCESS in 23m 22s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/4884e84edb7d4c61ba0b4c397172da4f : SUCCESS in 40m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fe424a6346384982349cbd1b7a143137966c42d9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:05:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d39f024d2b34d85aa8054bbef8dedcb : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"11c1d508692f817c63fb6a864c8331e459b3a5b1","revisions":{"11c1d508692f817c63fb6a864c8331e459b3a5b1":{"kind":"REWORK","_number":1,"created":"2020-04-27 09:11:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/723355/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/723355/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/723355/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/723355/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/723355/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/723355/1"}}},"commit":{"parents":[{"commit":"0d847bafce3936832c4574c32e01d3b28745e345","subject":"Fix seed VM provisioning on a remote seed hypervisor","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d847bafce3936832c4574c32e01d3b28745e345"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-04-19 13:30:36.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-27 09:11:00.000000000","tz":0},"subject":"Use upper constraints when installing Tenks","message":"Use upper constraints when installing Tenks\n\nBackport: train, stein, rocky\n\nThis fixes issues seen with a-universe-from-nothing using stable/train.\n\nChange-Id: Ib477de5f3af2e4c182d0c2999c274dbb5553531c\nStory: 2007572\nTask: 39469\n(cherry picked from commit 27779992b196201518da4cb220a0d29bb56cf307)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11c1d508692f817c63fb6a864c8331e459b3a5b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/11c1d508692f817c63fb6a864c8331e459b3a5b1"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
