)]}'
{"id":"openstack%2Fkayobe~723378","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I9323a38cb2bb627ff56f5713900be00595ea8d4b","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I9323a38cb2bb627ff56f5713900be00595ea8d4b","subject":"Remove nameservers with any IP in overcloud resolv.conf workaround","status":"MERGED","created":"2020-04-27 11:02:53.000000000","updated":"2020-04-30 11:14:25.000000000","submitted":"2020-04-30 11:12:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723378-1588245158863-81340386","meta_rev_id":"6f28c1bdaa0b39746ccd947a36bb99670fe520d2","_number":723378,"virtual_id_number":723378,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-04-30 11:12:38.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-04-28 19:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-28 19:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-04-29 13:23:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-04-28 19:51:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-29 13:23:05.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-04-30 11:12:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"51c2a3c512ef8296d274fb7736bf1b61edba9057","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 11:02:53.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e51b0a75ddce2e5940530c7c525a39d3d7576e25","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 11:03:02.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"420c0e11a5464859694210398ff0bfbfcf0d3813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 12:48:42.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e0a4b9209864401d814b3dfd8980e8fa : SUCCESS in 5m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/de9914130a774959ab0394a4c9d3cdff : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73d62992f3b049dcb77fb381fde705bc : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be8ecb20342478a9515fb33929cd9f0 : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e53063ce652b48b18e2e7b348dc523d2 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/657a014a57bc47bdbe2d775c765eb911 : SUCCESS in 7m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/811fe94a3afb41a5bb06f0d8c87c8f3d : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8db3712f219040baaeae317f55f86825 : SUCCESS in 8m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0fb701b43186463fae69bfdb5b4847b7 : SUCCESS in 31m 05s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/617fa83521cb4d4c9c3e0de06955d60c : SUCCESS in 1h 07m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f502af892ccf4071970c417944a2c184 : SUCCESS in 1h 14m 10s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a29b6dbef3614395a4ecc1862f0e9624 : SUCCESS in 1h 44m 19s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/fd787fc83966422b925ea015ae205799 : SUCCESS in 19m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ed263902461740babd2d3fc0664bf199 : SUCCESS in 19m 39s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0972b2f42db54433ae5653d30a340242 : SUCCESS in 33m 02s","accounts_in_message":[],"_revision_number":1},{"id":"6467fe523a125af79857127c691b590ae45f8a5b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-27 13:06:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"08c50ea221def9105636ee7e3e5fe8a3332ca6bc","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 13:29:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"df858f7e4bf5fc9053791b60de58091a66de4f9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 13:32:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9c5d49d0629c023646d9b94f31387a35d8839a5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 15:34:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/3419d6c614b34bebb95b47551c607261 : SUCCESS in 7m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/638828b03202429994e90cc60c276353 : SUCCESS in 7m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aaef5baa28cf42999833c949ad28d988 : SUCCESS in 8m 28s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ef314fa95134c9da4d4607f67f4f6a7 : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff988cba64db4be78004d26d3a3c4df4 : SUCCESS in 10m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e2dba551a95d43a19daae01e0213990e : SUCCESS in 7m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3257c6e69ed459cab398aa6f1afc6e7 : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f404cbb519df45d78c613743c7bf60ba : SUCCESS in 38m 35s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/2f05bd1a08904948b665ac47594820b2 : SUCCESS in 1h 10m 19s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ee3011dc355f4bd3878bcf03b0fdd610 : SUCCESS in 1h 04m 48s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/5121dca91f314ef8b203e54237251c78 : SUCCESS in 1h 43m 29s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/d76dd033a3214e8cb69d48f00d679118 : FAILURE in 25m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fc38f5fad1324dd0b5aa5201de669fd0 : SUCCESS in 23m 31s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/99db7d9540d745bab7ad9b924d6300b5 : SUCCESS in 41m 38s","accounts_in_message":[],"_revision_number":1},{"id":"0095bd46ae201b1e0dc7ef3db0c5174f05eda3f2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-27 16:06:54.000000000","message":"Patch Set 1:\n\nrecheck\n\ngit opendev issue: Failed to download remote objects and refs:  error: RPC failed; result\u003d7, HTTP code \u003d 0","accounts_in_message":[],"_revision_number":1},{"id":"b52e84e4c3febb3a339f81bef5bdafe9193a8cda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 18:08:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6e0cf2925783465ba4538d76551d1093 : SUCCESS in 8m 25s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a40650b2e2844ab48404ad4ae804a315 : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c951af9bbdb146619fae75a0b018db2b : SUCCESS in 6m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85643446fbf44a9ea4702b3af09a188d : SUCCESS in 5m 20s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/176cbd661c8649fbbd18f58116db47fb : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb39bf9ff214ecfb65829e4bb41101b : SUCCESS in 6m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2af5f383ceab46c9bd48560bab5a7543 : SUCCESS in 6m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9287626386b449fb9052537a6cb0e91 : SUCCESS in 11m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/05d522d458cb45d3ae276c1dd481533d : SUCCESS in 40m 24s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/9dc8fc91c8464fedbd38629e431394ab : SUCCESS in 1h 14m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/fb3bddaab39647a3b062214348e45eea : SUCCESS in 1h 07m 44s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/cfe411f062ec4d69adc025c748652d2a : SUCCESS in 1h 51m 09s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a42475a761934f838ef82f914d3e40b2 : FAILURE in 22m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e7e218d6f16a460fbb56cd5ea1724f32 : SUCCESS in 18m 54s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3f2e29360c2f4bc9acf5316c3548280a : SUCCESS in 46m 17s","accounts_in_message":[],"_revision_number":1},{"id":"eaa4040eefcddc097c4dc7e1a3efde021bac5050","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-27 18:40:58.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"2ac268649af88af21f18a7a3a0d542374d51ee98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-27 20:34:30.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0c3d1ade1974b19934212cf458227fe : SUCCESS in 5m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a5bda108b3e04ac0b7f74c329299b8d5 : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/54cbbfb2de664a4a8cf842dc8f020a7d : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0dae8e4adf4848cbb9a0e500ccd193c3 : SUCCESS in 4m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/35bd86982db94b5bae97b17fd116f69e : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b81f09e2f7c479b9af638dd631d046e : SUCCESS in 7m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/da618e986a054d7caffadb3b51c02bd5 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b2233cdf925845e18691855d3841e732 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/847cd606084f4aa6a8ccad919b559b95 : FAILURE in 8m 48s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/701950638e9f4cf8a65ed013f59fbf83 : FAILURE in 54m 59s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3f807e5922374a48ad2f6608791b00b7 : SUCCESS in 1h 17m 49s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/50838e4d57024b198562b22ccdc865a3 : SUCCESS in 1h 49m 36s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/0834f42b47554bab9be8d1989e633cd5 : FAILURE in 23m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/594614a972774bd28478a2116e7ccbd9 : SUCCESS in 17m 31s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c1056345c58a4b53a6b43a5fc21c99c5 : SUCCESS in 31m 39s","accounts_in_message":[],"_revision_number":1},{"id":"bc979ad27c43ce4039ca8dcd37c99d48d624880e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-28 13:53:51.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e00b05dadbcc732e4912ded54874cb15d7dd7f8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 15:58:17.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/02860055d1954076897a72503e912d77 : SUCCESS in 5m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b8b4ea134d67494b80e376b5b54e7359 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfdb57e6ee8f449ba508c3b77a7944e4 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3fadf96670b4a04afa203b5967f4cbe : SUCCESS in 4m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c3bd48c3730d41d48fd467a9f851e453 : SUCCESS in 8m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c1c66881f8b4d31bf11fd06bc359536 : SUCCESS in 11m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/40aebbc8db284668b0f846a9cbceade9 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/99bde3f71f1f4ac0906e7c95547d9f22 : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/790a67c73cd54b2bb03e86fc0034a820 : FAILURE in 8m 50s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/13d5751d603c491cb83dcdb71b18ce94 : SUCCESS in 1h 39m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b9513fc807644d849a2af46991abbf06 : SUCCESS in 1h 30m 55s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/20d9cc1411034b3392cc6f8e69d599a3 : SUCCESS in 1h 58m 13s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/80ff7c15360040e8b399bcce28d01bc9 : SUCCESS in 20m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c04bf3b56b6b476bbe8d0832514bd498 : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/bb7ca70eaa544770ad7a3745162d80f8 : SUCCESS in 40m 09s","accounts_in_message":[],"_revision_number":1},{"id":"fc1377abf7bd90f873bfc628b958943ef85e95e2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-28 19:51:48.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"2ece49d73b29749668704e068527c1687058efda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 21:03:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0de7058bbfa7478f9738d68ff30f31f7 : SUCCESS in 5m 28s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/92d034c5208846d49f5e204f56ef9b8d : SUCCESS in 6m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d59689cb3c21450a89b0cac9cb7b3ae4 : SUCCESS in 5m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69404242351a4a419cd9abfa56b6813a : SUCCESS in 5m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/16e9c853cb114d50b70d4ff559b6d7a1 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74bdda4440784f9cbd45bcc39cc42fa6 : SUCCESS in 8m 38s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d1908e52098c407eb1f0ee0e2ce01962 : SUCCESS in 9m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2c96a6589a424f22a78057200b28cedb : SUCCESS in 14m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/cda9dd779f3b46509332a32bfdea3498 : SUCCESS in 31m 41s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/8452537dc821408facfc39f0f36572e5 : SUCCESS in 1h 08m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1fe17d0fe20f4de6ad19c72318fa6e0d : FAILURE in 9m 37s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8b7ca43549244351aaa1e636384f96cb : FAILURE in 11m 11s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/05bd7df0f3da448a9481e56d78808fcf : RETRY_LIMIT in 8m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2d4dd75edd2041cfa647a55fbd406fb1 : RETRY_LIMIT in 4m 46s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e6e5432f04c744b9a4b894706c527ed7 : FAILURE in 7m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ea6152ba9e016b213d9a4c37703c2bc58a2b344a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-29 09:41:51.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"3f8c7ae783a58b290206f56efa7a429898a105ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 11:03:29.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca8ceea5fd5c4da595790b5f32f80934 : SUCCESS in 5m 17s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5a7a1c25e41a43e3bf7b58b484516c3e : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/11dcfb13a604465f9122740534bb6ead : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdc8ce1a9b24432abdf6af80283b8c58 : SUCCESS in 6m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ae84beaa6e4d46f781ccf4a9bed815ce : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/254cea70d58744d783fb8e82fa54f626 : SUCCESS in 6m 49s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/084ebe1927684b90813fd96987256fb6 : SUCCESS in 6m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0a02c379497d4f74a8c3c061b71388c1 : SUCCESS in 10m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d979d9ce665d45a8bf5908675aa44b56 : SUCCESS in 36m 32s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/a70d2de3fdc248779e0a45ba944bf8a9 : SUCCESS in 1h 07m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/05f98a923ccc4bf587921e16f7e31def : SUCCESS in 1h 15m 58s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f3939ee7bbb2479ba0c6a6c5db04c36e : FAILURE in 25m 04s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/1341ecdd71e54647b94e04725943b522 : SUCCESS in 22m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/901fb4dbe4f74916bc4ddb39fb29e965 : SUCCESS in 19m 50s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/62889baa59214490b4867553725614fe : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b0404164eee5a9bf65b3e849e583b4a4282e36c2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-29 11:06:00.000000000","message":"Patch Set 2:\n\nrecheck\n\nUnknown error message: error pulling image configuration: received unexpected HTTP status: 502 Bad Gateway","accounts_in_message":[],"_revision_number":2},{"id":"2a2cbabef903c8c77faeb0923564ba5679df8156","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:08:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7d898418960c42048842622e49f7eedb : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/d73b6658bbb24f5da8374488b91d7796 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a2281372ff764886a7b871dce626b063 : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a51c51969b8542de9b2a70af31542585 : SUCCESS in 4m 52s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32d198a6eeca432daf4f911412845fbb : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5bc4d80239094e7fab17b5265856fe41 : SUCCESS in 8m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c2c073f96742471da3418d0d65a0a642 : SUCCESS in 5m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fd9bcddc98464f438a57b55c26b5faef : SUCCESS in 15m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/42163a93278b4032b0ebcc6d2a78b3f7 : SUCCESS in 32m 24s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b2638461d4b44b7c84feb49218e8aa85 : SUCCESS in 1h 10m 36s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3251702b455f4287983f36e42e6bc04a : SUCCESS in 1h 19m 17s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/87109f3bdc1343b7a1a621bf932631be : SUCCESS in 2h 00m 14s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/3f282c28bb8b4cd0aadce0ea7df3e559 : SUCCESS in 20m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/cfc4c25b0082469e8aaa058ea653c63c : SUCCESS in 23m 21s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2691d404562f48eaaa3d37f28eb741c1 : SUCCESS in 37m 19s","accounts_in_message":[],"_revision_number":2},{"id":"c0fe679045b44e9e744db9314de9223c3f18270c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-29 13:23:05.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"aeb33387316728df2bdbbf12a10eb8b46871addf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 13:23:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5492613202299109f218459fdb0c6d10bfecead2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 14:16:10.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/de60df08ce5c488b9df9f09ddc1818ef : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6f2b2dbc31ab4822852be758ba4a5975 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47878124bd6848a5bb2993cf018c10ac : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8ff634a27eca437881283f08e9b6b59a : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/430d197fb5d24e939140e6eab80f5113 : SUCCESS in 6m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23353ccb5a7c43e9b2498376861b46fc : SUCCESS in 5m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c6749d24438e4f72b6e271c5366930c8 : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9823c6be055a4f1293951f33beb3d705 : SUCCESS in 32m 23s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/004d5dafbe494def940af47246a42ee8 : EXCEPTION\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5ac0fb94e7644dbd9076c65349e85088 : EXCEPTION\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4b6fc8fe730741b19a2a507b158cb164 : EXCEPTION\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/f999a1f4bd0e46e19cedda15679c1d03 : SUCCESS in 21m 48s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6f637cda9ccb4d92ad0fc02d66f0de34 : FAILURE in 24m 09s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/aa552fc471d04243b58836082448759e : EXCEPTION","accounts_in_message":[],"_revision_number":2},{"id":"a898e3d9526ca7e50c167129f22ecf5fc535b2fd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-29 14:36:58.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f20a462b388e95ee9f45d64a334d2dc04d7667c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 16:48:42.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c0f89549805d4e4781f3e06d40b1a08b : SUCCESS in 7m 04s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/030cab951d0348ee948b14bf36a88a02 : SUCCESS in 6m 58s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/42bff7968da54968a44f45d5515c6e60 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/668fcf0a3d054cb3bd34d5c1d761a7cf : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4c1e3390b2a841c0975cbe8d087ff969 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d96feca700a84e3180456c63947f5177 : SUCCESS in 9m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9e844bd6fd254717872e40402900ee8e : SUCCESS in 7m 08s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cf043c0edbd643b2a91f8c28793e0307 : SUCCESS in 12m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/99bed68cc7a845078fbe629ff7e42115 : SUCCESS in 38m 46s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/46bcbd7248e748f9898dc6587c66dfe3 : FAILURE in 1h 06m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4859b8bda20d4baeb8a210fa933a3053 : SUCCESS in 1h 14m 10s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8d0f5e2591354361a6389418822975f1 : SUCCESS in 2h 04m 25s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/880b9d6fad53481d8b46fd11be509385 : SUCCESS in 24m 16s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bf1bfff5d5404ea192797a8093652e25 : SUCCESS in 20m 31s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ebb2ae7b2a5549d69976f9e83aed0b4f : SUCCESS in 36m 04s","accounts_in_message":[],"_revision_number":2},{"id":"740b273e860024e14deb71a5892c1d0f99c9a0ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-29 18:30:17.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"a3327a15b7e4737028fa57d5f074221fcc52fcaf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 20:40:58.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cd6d6a56498b44e693d466782a89c12f : SUCCESS in 9m 01s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ca219470d1d247e9863721ceb94e7655 : SUCCESS in 8m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1780d8e445b9455b8bece732194ace31 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5003a325f8634770806082791ba4b701 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fa17e469693b44848dd3bd606608971e : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2f3ec1c80514883892485bbfc0a8424 : SUCCESS in 6m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ad76a3754f94b9b976ec660dbe36cd9 : SUCCESS in 4m 56s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/060c3d6a6ee74bd69beb7c59ee53a2e5 : POST_FAILURE in 24m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8749c495498941b28c9e84064869824e : SUCCESS in 37m 23s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/67e2f8381c384174a342ce13b6086daa : SUCCESS in 1h 08m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ba47e9383f954f4d88289d4e7dcc9a01 : SUCCESS in 1h 09m 48s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b05c9bd303c848599890e8bfeae993b2 : SUCCESS in 2h 01m 17s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/faee1c0adf0d4a8390ef687c2f7e898a : SUCCESS in 24m 17s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b54ca6e951bf43b38c975ad7f57b887d : SUCCESS in 20m 00s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8871be5e93ef4dea88ca43166f34123d : SUCCESS in 34m 47s","accounts_in_message":[],"_revision_number":2},{"id":"f6107aaf668b639c060ef9244db49726ad21ba11","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 06:55:26.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f8aac7ff9bda88805ef6b482016e21254012e61c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:51:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ccdb41d32fa0445b9543cdbf3b70ae8e : SUCCESS in 7m 32s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a4bf096ab79d421fb75e0963c72adb66 : SUCCESS in 7m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/eb48ae2e06cd43988295f9c1bffc44b9 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc994fa8fa324a3bbaeb11cdd5c01560 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/7ca9e636334049aeae03b3a291305b55 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/484880c077b845e4ad4aef878a736396 : SUCCESS in 8m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e136d81175084528ab13f65a4696d474 : SUCCESS in 5m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/87d7e68da6c7428aae540e8d31e607d4 : SUCCESS in 10m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2009cc5dcecf4a5198eea5007506f2b8 : SUCCESS in 30m 23s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/7eddb76a98a94214aeb544b14d77b407 : SUCCESS in 1h 13m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a945e94a11db4dddba052f7ad0c8a494 : SUCCESS in 1h 07m 06s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/aca0c7b61d124e479a5c538f8479dcc4 : SUCCESS in 1h 47m 51s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/db9b61327dd54ef98cdacceb8ccfe261 : SUCCESS in 24m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ccd9a8fdc46f4bdf8776eadbcdcdba39 : SUCCESS in 18m 55s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b3705fbdaf75474cb5413fbdc3fb4dea : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":2},{"id":"d922713eae35b7b28c0c8b38617ea3d0609e272f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 08:51:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1046e6df09e656a6dc85a53a2a87ea329fe9b8a8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 11:12:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/42953b5158484111bdbda53f5a900156 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6e813b4d54554f708431d2612b08a791 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba5b302d15094fd1a44416bc13439ef4 : SUCCESS in 4m 27s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/209eea91573243d6865cc8e8bcd2b2c3 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ec2ccb594264f76b9d6040f29f1bca7 : SUCCESS in 6m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3acb341bb9414dff8aa47a50783caf15 : SUCCESS in 4m 40s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4012e7d29829458f91dc0bd9e4d748bf : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a1befb4d8668436db96191daed454a92 : SUCCESS in 43m 19s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b7f584ee2a414a92bca20977304703e3 : SUCCESS in 1h 07m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8f72cbc06aab4897be41054ac344cea5 : SUCCESS in 1h 16m 40s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/47bb70a477bc4bb19f313d057f397f7c : SUCCESS in 2h 16m 43s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/261438283a3d48e0ad09833237349ae9 : SUCCESS in 22m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dd58b8a58c364bc9a7def3d5a1074a26 : SUCCESS in 18m 11s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/ccd36171dedd46179feb9b7ad1d67bec : SUCCESS in 36m 05s","accounts_in_message":[],"_revision_number":2},{"id":"144174ebc6c307b40ecd3e5b60fe7846967a9c1d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 11:12:38.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"6f28c1bdaa0b39746ccd947a36bb99670fe520d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-30 11:14:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3ea362a907848a69833728bb527611b : SUCCESS in 1m 30s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9f6e4395390fa8a4f2441cf9a001e4ea5849d918","revisions":{"12257df33562783f3267bf3ebc20d635000942b6":{"kind":"REWORK","_number":1,"created":"2020-04-27 11:02:53.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/723378/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/723378/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/723378/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/723378/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/723378/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/723378/1"}}},"commit":{"parents":[{"commit":"04670de47bf362ff5dc52511b2e50997a5aaa652","subject":"Merge \"Fix passwords.yml generation with vault encryption on Python 3\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/04670de47bf362ff5dc52511b2e50997a5aaa652"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-20 14:59:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-27 11:02:53.000000000","tz":0},"subject":"Remove nameservers with any IP in overcloud resolv.conf workaround","message":"Remove nameservers with any IP in overcloud resolv.conf workaround\n\nKayobe has a workaround for CentOS cloud images which contain a bogus\nnameserver entry in /etc/resolv.conf. By setting\novercloud_host_image_workaround_resolv_enabled to true, the entry would\nbe removed. Previously we removed a specific IP address - 10.0.2.3 -\nthat was present in the CentOS 7 images. However, it seems that CentOS 8\nimages have a different IP - 192.168.122.1.\n\nThis change fixes the issue and becomes resilient to future changes by\nmatching any IP address. This should be fairly safe, since this\nworkaround is opt-in.\n\nChange-Id: I9323a38cb2bb627ff56f5713900be00595ea8d4b\nStory: 2006574\nTask: 39484\n(cherry picked from commit dcd5159c172f50b9708cb146e4f3a11060dee243)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12257df33562783f3267bf3ebc20d635000942b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/12257df33562783f3267bf3ebc20d635000942b6"}]},"branch":"refs/heads/stable/train"},"9f6e4395390fa8a4f2441cf9a001e4ea5849d918":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-04-28 19:51:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/723378/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/723378/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/723378/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/723378/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/723378/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/723378/2"}}},"commit":{"parents":[{"commit":"5f836518cfe964d2733f5a7db327fbfad3b8e8b0","subject":"CI: fix kayobe-tox-molecule job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f836518cfe964d2733f5a7db327fbfad3b8e8b0"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-20 14:59:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-28 19:51:48.000000000","tz":0},"subject":"Remove nameservers with any IP in overcloud resolv.conf workaround","message":"Remove nameservers with any IP in overcloud resolv.conf workaround\n\nKayobe has a workaround for CentOS cloud images which contain a bogus\nnameserver entry in /etc/resolv.conf. By setting\novercloud_host_image_workaround_resolv_enabled to true, the entry would\nbe removed. Previously we removed a specific IP address - 10.0.2.3 -\nthat was present in the CentOS 7 images. However, it seems that CentOS 8\nimages have a different IP - 192.168.122.1.\n\nThis change fixes the issue and becomes resilient to future changes by\nmatching any IP address. This should be fairly safe, since this\nworkaround is opt-in.\n\nChange-Id: I9323a38cb2bb627ff56f5713900be00595ea8d4b\nStory: 2006574\nTask: 39484\n(cherry picked from commit dcd5159c172f50b9708cb146e4f3a11060dee243)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f6e4395390fa8a4f2441cf9a001e4ea5849d918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9f6e4395390fa8a4f2441cf9a001e4ea5849d918"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
