)]}'
{"id":"openstack%2Fkayobe~724397","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ia24e328d4531201d76a65b6385e4463bb1f3c5c6","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Ia24e328d4531201d76a65b6385e4463bb1f3c5c6","subject":"Fix ironic inspector rule creation idempotency","status":"MERGED","created":"2020-04-29 18:25:40.000000000","updated":"2020-05-18 11:25:39.000000000","submitted":"2020-05-18 11:25:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724397-1589801139175-ec73deff","meta_rev_id":"eef6affe9b6db271161956a8e68154770e440356","_number":724397,"virtual_id_number":724397,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-18 11:25:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-05-18 09:24:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-04-30 06:56:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-05-18 09:24:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-04-30 06:56:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-05-18 09:24:30.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-05-18 11:25:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"07268eec6fe3fbd6671bd7b047edce87dacf7044","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-04-29 18:25:40.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8f30bd37ee68131980458e519512a887d01845a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-29 20:31:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/156c03e71e1f40beb6df147c3503a2bb : SUCCESS in 5m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/af08e0cc11284cc1bc4862b6b001a849 : SUCCESS in 7m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8d43ae74607a4d95969fb475d7c99ce6 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9aa0b4966404ee1931296145206d54f : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/04126e55635b4146846d6d993cc2db70 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de99b52a0e034bd4b93f5db5d073b269 : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/afc031f35c0044a5a24da2c202fb1539 : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c3a2eb70b7214921b91e5b112ac8bc57 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c87281d611fe49798aebee36569cd089 : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/70fbf1492031459793d5086e76bfd781 : SUCCESS in 29m 44s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/3b1605705d12428e9dc6224da85ae3b9 : SUCCESS in 1h 16m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0a99279f1a0346b7b7ee515fa0b7e30e : SUCCESS in 1h 11m 04s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6576c10712be4d8f887ddde147e9c397 : SUCCESS in 2h 00m 20s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/67a3c43a6cbf4aaa811407a93ed3df5f : SUCCESS in 25m 59s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9bb8b316df754c66b0e9f64db9abb110 : SUCCESS in 20m 33s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/598b6292d9344f329c36480966407d7b : SUCCESS in 45m 46s","accounts_in_message":[],"_revision_number":1},{"id":"866e6297527f56f2ad2a44a9800918bce2cba139","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-04-30 06:56:34.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"cb20402aa2ca2122a52d1ce4805ae53fac9822f3","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-18 09:24:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"374daf9e212e20eb1e803f560d2919d2a8a16d77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 09:24:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"028ca5670a533946ae50f75f24ea5e6eb146831e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 11:25:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5e29d9f70cc5444baccb1b7899523c87 : SUCCESS in 7m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb9c0416a9154c9aba33e9ebb5446fc8 : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/690cd0840c6a42ada485faab55510a65 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/dd6d51604fa84a40a1d79ad83e1e30d5 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e1eeb5ae0f4e4811ad057691488a7669 : SUCCESS in 8m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8bc75a94269641bfaf622ea3e68e1196 : SUCCESS in 5m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/91f2eafedd0c4703b0b4aea7a41f426d : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ad858ffead664cbca1dc76298da0f83b : SUCCESS in 10m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08fc8c39dc0543bb9c743b190149c47d : SUCCESS in 34m 24s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e73a7f2a44fc418bacc49bdb15c52fd2 : SUCCESS in 1h 08m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/686f03e5bee1424ebab71f906b7391f7 : SUCCESS in 1h 10m 10s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/190b4e68b58a48d196284baf558748d3 : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4039f01eb8854bb3be0e8bb7db69f822 : SUCCESS in 19m 12s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/46aa9f67966846e782b98b2348670ea1 : SUCCESS in 1h 55m 43s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/2a2b358b7fff454ca377d4a4bfb14d49 : SUCCESS in 22m 38s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c1a1342bffc246369b8e78199cb11ef9 : SUCCESS in 22m 58s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1d152c7217e44e70aacedf9d1ae8ab5b : SUCCESS in 41m 20s","accounts_in_message":[],"_revision_number":1},{"id":"eef6affe9b6db271161956a8e68154770e440356","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-18 11:25:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"64de1b4291434034053cad1fa64623a1191f1714","revisions":{"64de1b4291434034053cad1fa64623a1191f1714":{"kind":"REWORK","_number":1,"created":"2020-04-29 18:25:40.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/97/724397/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/97/724397/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/724397/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/724397/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/97/724397/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/97/724397/1"}}},"commit":{"parents":[{"commit":"9d644b12b0a8f31bcdb1db11fb9e96f8df1687d9","subject":"Merge \"CI: fix kayobe-tox-molecule job\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9d644b12b0a8f31bcdb1db11fb9e96f8df1687d9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 15:55:14.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-29 18:25:40.000000000","tz":0},"subject":"Fix ironic inspector rule creation idempotency","message":"Fix ironic inspector rule creation idempotency\n\nIronic inspector rules are registered both with the seed and (if using)\novercloud ironic inspector services. These tasks often show up as\nchanged even when no configuration changes have been made that would\naffect the rules.\n\nThis is caused by inspector returning default values for fields that may\nbe omitted in the requested rule. This change fixes the issue by\nincluding those defaults in the comparison.\n\nChange-Id: Ia24e328d4531201d76a65b6385e4463bb1f3c5c6\nStory: 2007399\nTask: 38997\n(cherry picked from commit 2a00b4cc67df91c5eac10f4762aff543d3740e05)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64de1b4291434034053cad1fa64623a1191f1714"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64de1b4291434034053cad1fa64623a1191f1714"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
