)]}'
{"id":"openstack%2Fkayobe~724963","triplet_id":"openstack%2Fkayobe~master~I43bbe9dae3d6796e308fbf66cb04d16b57ff5e37","project":"openstack/kayobe","branch":"master","topic":"story/2007612","hashtags":[],"change_id":"I43bbe9dae3d6796e308fbf66cb04d16b57ff5e37","subject":"Install coreutils package before using configdrive role","status":"MERGED","created":"2020-05-02 07:18:14.000000000","updated":"2020-05-14 13:21:04.000000000","submitted":"2020-05-14 13:06:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"724963-1589461607548-30010a4d","meta_rev_id":"e77ab3442c3ae17fa0c2a7ece7d523566190857f","_number":724963,"virtual_id_number":724963,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-14 13:06:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-05-13 20:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-05-13 20:30:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-13 20:30:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2020-05-13 20:30:08.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-04 16:11:42.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-12 09:48:55.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-05-12 09:56:39.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-05-14 13:06:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2caf587cc4bb9d92da1d5ffa18296608cc39e5cd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-02 07:18:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e7b7c8e470f4054f4c7d375ed0b3c943854948ad","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-02 07:40:32.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"98682fb01e43224bc58796e89f560598b1819ccb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 09:33:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/413c0b5502ee4100a99ae75380300540 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4cc851f09241491caaced8111f90d605 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d6e3ef02dc804a2b8e800bd30cbfd281 : SUCCESS in 4m 39s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6bc5dda71c654150b2b08df6f3b199d9 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d61f46c06bd54f3d9bd201db05e3e6b3 : SUCCESS in 4m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dc1f98ef4fd4b849cf9126805264d0b : SUCCESS in 7m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b8d5e000e244487a9ad32427d7551670 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/457969a53c954232980e2e97929b48b2 : SUCCESS in 8m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2a731a781ca34a57afe7b7a9e6011842 : SUCCESS in 28m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7987edc2b01f4608a041411e454fdaf2 : SUCCESS in 1h 04m 46s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8cfddc1ce3184c928094c9996f78a3de : SUCCESS in 15m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cc0b61bfb9014c1886baf2cd3fd1e405 : SUCCESS in 1h 50m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9c1d049579734221a9216bbbd152d202 : SUCCESS in 18m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4c3b4e2b06a64bf3b1451fc2c7ebf292 : SUCCESS in 25m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ebbc2830123ab05dcda5679fb0879f162618555c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-02 09:51:58.000000000","message":"Patch Set 2:\n\nBackport to Train too","accounts_in_message":[],"_revision_number":2},{"id":"1450b78c411d0e41a0eb7bb4a8b7f11a476dd293","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-02 10:02:50.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"50974ae17a6e0658be09bcac0d0ddd08b1e2fa5d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-02 10:10:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d4c0071351654786d57bf2180e3b449d263b48bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-02 12:01:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d057c8fd641c4684a27372011aefefa2 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1efdbcb17d904db08bdf0c3dd65db437 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e7bfceb8e3a74d24bd03147b4fa084b0 : SUCCESS in 4m 54s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d6be6eec1c514674b28d0320c24b8caf : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3acd350c653941e3947ec8b6f5f099de : SUCCESS in 4m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/665cd4a71fe747028bea9f045fb47506 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/74dc2447b0cc4f88afbdf55f7704f065 : SUCCESS in 4m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23afa26bd4154a3e9e35c8d2ae064591 : SUCCESS in 4m 20s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f38930be68b54c009c24336623176fa1 : SUCCESS in 9m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6e17721c184c4146a9c4a20b71858079 : SUCCESS in 35m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8a91cdb72b7f4bb28db035b74bdd59f5 : SUCCESS in 1h 18m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/86682e9a42814e6996f5f7ab64b12fdd : SUCCESS in 17m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/851f5e5e52ca44bf9ef7e272ea2d6da8 : SUCCESS in 1h 49m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/30c8ab76a2c94e3f8708be7f14954924 : FAILURE in 10m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/553693637af64681a23fcb0dfc29b007 : SUCCESS in 25m 30s","accounts_in_message":[],"_revision_number":4},{"id":"4c165d0e0d154cbbfccc7efa39b8f97c1b2843ce","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-03 14:34:03.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"acfa724cad9b864164bc31aafcf0368b0180a9f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-03 16:28:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/79db96e68cff42d68d7df8865f7789f5 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31bf6b8a208c4be192d15f3b35a42a12 : SUCCESS in 4m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/922c78e08f6d447c9165eb9880825616 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2dae222206ca4d52ba544740baa727f0 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/631e7f5b1d3f49379bd9e0dbc964189a : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8049fd418664d9ea27f4c30315f02d1 : SUCCESS in 8m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ffb32a992efb41deb9e30130fed2c668 : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bdf4cf55c6c04d27b1d257e0ce368895 : SUCCESS in 4m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4c9bbdaa431e409aaaf6245600c7cbf0 : SUCCESS in 8m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ee00be6a886462db01decdcb1fc6981 : SUCCESS in 34m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/633b74e7079a48a9b01253dcc1bca5c4 : SUCCESS in 1h 07m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0dd1934f809b4f87a81f1b2c9d1587bf : SUCCESS in 15m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0617c9b065d54717b5bc98d76fbbb074 : SUCCESS in 1h 53m 49s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a459d6fd83a84750bd07506f1fc64eb4 : SUCCESS in 18m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b15917c41b5f49fdbbb835b563aa2f3d : SUCCESS in 24m 55s","accounts_in_message":[],"_revision_number":4},{"id":"f3d7d206bde4ccc2cebfa8673d79c8162b8a1e94","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-04 10:12:28.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"9f529762c0abf73da30cdab356934fa13b6e4580","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-04 16:08:17.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"77fd31f4150fa5f0562e8dfb08d9d7e7928a6d5a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-04 16:11:42.000000000","message":"Patch Set 4: Code-Review+2\n\nI suppose it won\u0027t hurt.","accounts_in_message":[],"_revision_number":4},{"id":"fbf79720721e0917920bd1a6abe92d7279b761f8","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-12 09:56:39.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"ff718ab13aa61ec0d1091cb3da91956902f355b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 09:56:50.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"c7a8cc1246c07ab9e8bbb149a8b5f5483a22f71f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 11:25:56.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bca869ee27504ddf9c8e23286df5219e : SUCCESS in 4m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c28616a732be44dca6edeaf467224dc5 : SUCCESS in 4m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b0bbbb151d3242cb9bf9e36bfa52b889 : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3dc2aedf11c491a9dd19718cab98d97 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2427b50c119a4608b70f06f332184ba8 : SUCCESS in 3m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/226a7c3301ab4b2394f4d0d056b31050 : SUCCESS in 4m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/45a8aefa98aa4e4f84d2f673c475ecb5 : SUCCESS in 7m 57s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2193de805d184826b119d19471f30e0d : SUCCESS in 26m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/426a0b87682343988e6a6182df56c7d6 : FAILURE in 28m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/787cb21be1de49fd80dcd02b7b7d53f6 : SUCCESS in 18m 06s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a751a6a0c8d9465a8168ef6f4c92342a : FAILURE in 1h 27m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/56af2cf9fb044d29a311d986f9bf8293 : SUCCESS in 17m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/750a21df178c41ad96aeadf3dde3ae35 : SUCCESS in 26m 27s","accounts_in_message":[],"_revision_number":4},{"id":"bf2717c54e352cc135d0f4c85c5ad698401c3b48","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-12 12:28:26.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"a3fabb0515d2e2fdfeb0964c034d183d9d5cd66a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-12 14:41:03.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2b7c09f65b74683b1198046af255d28 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e940c4fccfab47e79dcae70a9206f194 : SUCCESS in 4m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4d8a16a06d8a4911a4c91607c605b074 : SUCCESS in 4m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6dd108da0c4244ea8a3f0540c02a5248 : SUCCESS in 5m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1cda67440b8f4c7c95f3751082ddc62a : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cce95dc00df4e8ba4194347cfc38d81 : SUCCESS in 8m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cf3ddf40a674410828d6dc06b9ecf7e : SUCCESS in 4m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4de1cb9eb4574ee2bcc70203d394a83f : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3b09c3ac13fe41f096ed357ee0466a85 : SUCCESS in 9m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3a4336a821ff453e91d3cbd385d5b34a : SUCCESS in 34m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1cd8c064b0e9466983b7c7964b6a9980 : SUCCESS in 1h 09m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9c7efb2eac264c8fa30e13fc26c3a005 : SUCCESS in 16m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f3230dec72ce4947b99b192d309308ab : SUCCESS in 1h 59m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8e13e8fb88e04f678dad407cafb936b0 : SUCCESS in 18m 19s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fbf83b6253734ec4bdc5584662a00eec : SUCCESS in 31m 14s","accounts_in_message":[],"_revision_number":5},{"id":"e3483f283fc255a5c7046eab2818291e85729fef","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-13 09:21:26.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"356c06ea2610a27ad12325022621f0c9978ff65e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-13 16:34:47.000000000","message":"Patch Set 6: Patch Set 5 was rebased","accounts_in_message":[],"_revision_number":6},{"id":"13e652fb7700589744f5e7e00583d1ea7a22d9bc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-13 16:34:57.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"6ace14099e110e46e93dd02b15f4d1de6fcfd023","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:02:53.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee537b76ac92445db2b9ba39fefc48b4 : SUCCESS in 15m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dbd5f53fdd9c4b6b8705e674653568da : SUCCESS in 14m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2bcadd25c54e4bcda2fc036f74895838 : SUCCESS in 14m 28s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/185dd7808d4645a68e6594957247ddb3 : SUCCESS in 15m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75eababadf6c4a84b2f3d7e00957782c : SUCCESS in 16m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c86904e1e50d4cc4afa03694c2e12f89 : SUCCESS in 13m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bf40e09570bb414cb61ecca82ff26da7 : SUCCESS in 12m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c26e6ecb26f345bda6d47d86c928dc18 : SUCCESS in 9m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c9264d62fd9e4bf6aa52e051e065aefe : SUCCESS in 19m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fd1d15887383443fb9fb61f12bf5c2ab : SUCCESS in 38m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/02909b3cf90148a6bc945fb587b8844c : SUCCESS in 1h 24m 09s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f5b18bb8190e4d808ae43a260c2a0945 : SUCCESS in 24m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8963d0a74f11492a960ad3ab81002e15 : FAILURE in 1h 15m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/191ce27cf3e04f07af243d56d73563a1 : SUCCESS in 30m 26s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/019e0693ebb64fd2bee2d38cf96dc157 : SUCCESS in 27m 59s","accounts_in_message":[],"_revision_number":6},{"id":"6986fba19fc92462df97450722eaa34c9fc8576b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-13 20:29:15.000000000","message":"Patch Set 6:\n\nrecheck","accounts_in_message":[],"_revision_number":6},{"id":"06ba6a09153fe8edd1e85085157e75d75254161a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-13 20:30:08.000000000","message":"Patch Set 7: Patch Set 6 was rebased","accounts_in_message":[],"_revision_number":7},{"id":"ace08dd649baf2e50c82d282abde24e5f6469c9e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-13 20:30:17.000000000","message":"Patch Set 7: Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b41aa0a34c249aa465fcefc7fd699b2cbaf57942","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 22:34:45.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4591ff632da64d76b513f0bfc32d7dd2 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bd7ce7d7b8043d9bbef507e2f3dbc43 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63f41b9996324207b7b6f23b8e8595aa : SUCCESS in 3m 56s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/41a03a75e1d940faabe0d589e9479023 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/991c5b0845a74cbba02a765dbd518ade : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74a1bf4d2983478789ced90f30b36270 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16579ebfb6b14eecaa236b62ec8af320 : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fc26d34f90414c94989e2368c862c482 : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4971473a039c4da2878d5442ce6341fe : SUCCESS in 10m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29239925754d4f53a13982be37373efa : SUCCESS in 35m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/aa2e67e3a6e441e8a8bdb4d91a849f54 : SUCCESS in 1h 16m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5e4d2e3f97a54096abf2d3775aba713f : SUCCESS in 15m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fe98f2835d6d41a784098571ac766a02 : SUCCESS in 2h 03m 11s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a24e1bb38b404ba79974dc1bbacd02fb : SUCCESS in 18m 06s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1b4ab939e6924529b017f28996d8dfbe : FAILURE in 27m 25s","accounts_in_message":[],"_revision_number":7},{"id":"c779b24a170937bf895b5fec41f772df0920cfe2","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-14 05:50:39.000000000","message":"Patch Set 7:\n\nrecheck\n\nIssues downloading cirros","accounts_in_message":[],"_revision_number":7},{"id":"56359be706db108191cf7450f51b8dc265cc884c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 07:52:00.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e11e5e0c626462daa46adaad4e41e9d : SUCCESS in 4m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f8a959d6d7e4dccadbf88fbda1ab469 : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16eb6b8c2e27419684245084c92a379e : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9d7f6765c0064d1b840f0370845b0dbd : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/07fcc46ecc3a4fd981e24869912ed6cb : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f790800d2d9742e295c1a80666ce62de : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/209e3e283f5b4337b507f96b3fc61c2f : SUCCESS in 3m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/924a4834033d4d1bbb7045eba2a097ed : SUCCESS in 4m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d1e63d5769094a17805f7a8622ef5337 : SUCCESS in 9m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d9aa4a401a3b4dab887b249b15111bd2 : SUCCESS in 34m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c968bcd13d3747f1ad2f76c4a62573ad : SUCCESS in 1h 03m 48s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/722d808c286b4d6685e944b433d48bfe : SUCCESS in 20m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3c048c1ef77d438485b8bd8442082071 : SUCCESS in 1h 56m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/db61045230e649f0961075418224e646 : SUCCESS in 17m 19s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f0e836d85e1e4146986015feeff563f1 : FAILURE in 21m 22s","accounts_in_message":[],"_revision_number":7},{"id":"51b418962c951f253425f17f3418f8d0d0c987b9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-14 09:00:02.000000000","message":"Patch Set 7:\n\nrecheck\n\ncirros seems to work locally","accounts_in_message":[],"_revision_number":7},{"id":"c391bbd8745160ec61ee12963e881c811de0b2d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 11:01:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/166f1cc6979b44baa88ecfb1e5874558 : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b4a67d29f3d444b89693945d2bb2eb9 : SUCCESS in 8m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9f4ed8c715a0425cb08b6ac5d2d0352a : SUCCESS in 8m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d0dec2f7b07048a39622c90d6c0d08ff : SUCCESS in 6m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/80ade6451a7942d888d6cab9271ba8ac : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/220bed25861542c8aea5ac311088b154 : SUCCESS in 9m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e4cb4be119840388cb956386d4923a7 : SUCCESS in 4m 43s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/551b8bd9f26e4c66ad37ab4dc9267b01 : SUCCESS in 6m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2562577a8a884f3c9788049afa22b9dd : SUCCESS in 10m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0428a0df87064cd9a54e1c44aad404a2 : SUCCESS in 35m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/95ae183effde44d5b21ee42b9746f379 : SUCCESS in 1h 10m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5259043d29ee42fea85e6b8d849ec06b : SUCCESS in 18m 12s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cd9aa3e71d7e4976a10a02c9bd4d7f6e : SUCCESS in 1h 59m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d09afbc6b778444eb49e8de3bf6e9760 : SUCCESS in 24m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f1abd19f6c4f4d74878e7e0412a09501 : SUCCESS in 27m 42s","accounts_in_message":[],"_revision_number":7},{"id":"ba431e411472cdc0f682e9c3cff09f3cd771c8ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 11:01:25.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"85d875050c57cc218f349fcf72aad5ecdca7e696","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 13:06:47.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0174fa9d0e704abbaba7ef44c746f69d : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd99fa1df6044018b24a818e2b4309bc : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4156b21c784a4d5b825c3a0073d40099 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48eb8fe1b13d453785a15174cffca69a : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97381c7846b045b2acbaf2c4385ed78b : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/658e4b0821714973af778788de58caa6 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/24f05214982e4d9eb1a4ef2473025340 : SUCCESS in 9m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7bfa7e6f871441319feda90f07e13b87 : SUCCESS in 39m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0f526f8c665344878231aa21625f3d99 : SUCCESS in 1h 07m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c547d0740c7645f39b63e8ed450289cd : SUCCESS in 14m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d0ac1d9b94884e6790f39fde5d8fea11 : SUCCESS in 2h 00m 34s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/73f52ec946cd40878182a6d25a99a33a : SUCCESS in 17m 59s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/632ae0727f624b28b01be6f34c4dd4e1 : SUCCESS in 26m 49s","accounts_in_message":[],"_revision_number":7},{"id":"4c30c5a884396c3802aa65352d70c6c56e104bf6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 13:06:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":7},{"id":"d88826f042163b9ae6bed89b23c4d63c9e727440","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 13:11:17.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dec0491d0774745a3b5f2e9e3805bcc : SUCCESS in 3m 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f74d2dbb2a844ff2a50e71e971825121 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":7},{"id":"af33430b686bd89db9cc741a503981f2da5dc7f9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-14 13:21:04.000000000","message":"Patch Set 7: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 6ef18d6a4e7315caf8e6cd353f91886585657634","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"fc017249c23cb819fd44202e2cff80d563cab8ee","revisions":{"aa0eeef8b2b4d64b583440d3a0e8811d0112ddf1":{"kind":"REWORK","_number":1,"created":"2020-05-02 07:18:14.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/724963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/724963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/724963/1"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 07:17:47.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 07:17:47.000000000","tz":120},"subject":"Replace coreutils-single package before using configdrive role","message":"Replace coreutils-single package before using configdrive role\n\nOn seed hypervisors running CentOS 8, the configdrive role will fail to\ninstall coreutils if coreutils-single is already present:\n\nError:\n Problem: problem with installed package coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8_1.1.x86_64\n  - conflicting requests\n\nUntil the role handles it, install coreutils using the --allowerasing option\nwhich will remove coreutils-single at the same time. Unfortunately this option\nhas just been added to ansible:devel [1]. Use a command task for now.\n\n[1] https://github.com/ansible/ansible/pull/48319\n\nChange-Id: I43bbe9dae3d6796e308fbf66cb04d16b57ff5e37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa0eeef8b2b4d64b583440d3a0e8811d0112ddf1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/aa0eeef8b2b4d64b583440d3a0e8811d0112ddf1"}]},"branch":"refs/heads/master"},"ff1203cc05b44899431d3659aeb2af75591add82":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-05-02 07:40:32.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/724963/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/724963/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/724963/2"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 07:39:21.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 07:39:21.000000000","tz":120},"subject":"Replace coreutils-single package before using configdrive role","message":"Replace coreutils-single package before using configdrive role\n\nOn seed hypervisors running CentOS 8, the configdrive role will fail to\ninstall coreutils if coreutils-single is already present:\n\nError:\n Problem: problem with installed package coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8_1.1.x86_64\n  - conflicting requests\n\nUntil the role handles it, install coreutils using the --allowerasing\noption which will remove coreutils-single at the same time. Use a\ncommand task for now since this option has just been added to\nansible:devel [1].\n\n[1] https://github.com/ansible/ansible/pull/48319\n\nChange-Id: I43bbe9dae3d6796e308fbf66cb04d16b57ff5e37\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff1203cc05b44899431d3659aeb2af75591add82"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff1203cc05b44899431d3659aeb2af75591add82"}]},"branch":"refs/heads/master"},"f8e0ec771c0505e88dca0066bf7a28d4012f045f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-02 10:02:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/724963/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/724963/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/724963/3"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 10:01:57.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 10:01:57.000000000","tz":120},"subject":"Install coreutils package before using configdrive role","message":"Install coreutils package before using configdrive role\n\nOn seed hypervisors running CentOS 8, the configdrive role will fail to\ninstall coreutils if coreutils-single is already present:\n\nError:\n Problem: problem with installed package coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8_1.1.x86_64\n  - conflicting requests\n\nUntil the role handles it, install coreutils using the --allowerasing\noption which will remove coreutils-single at the same time. Use a\ncommand task for now since this option has just been added to\nansible:devel [1].\n\n[1] https://github.com/ansible/ansible/pull/48319\n\nChange-Id: I43bbe9dae3d6796e308fbf66cb04d16b57ff5e37\nStory: 2007612\nTask: 39607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8e0ec771c0505e88dca0066bf7a28d4012f045f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f8e0ec771c0505e88dca0066bf7a28d4012f045f"}]},"branch":"refs/heads/master"},"9ff1a74f058cc496a74a298819a1b818beddd9e8":{"kind":"REWORK","_number":4,"created":"2020-05-02 10:10:30.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/724963/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/724963/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/724963/4"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 10:10:09.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 10:10:09.000000000","tz":120},"subject":"Install coreutils package before using configdrive role","message":"Install coreutils package before using configdrive role\n\nOn seed hypervisors running CentOS 8, the configdrive role will fail to\ninstall coreutils if coreutils-single is already present:\n\nError:\n Problem: problem with installed package coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8_1.1.x86_64\n  - conflicting requests\n\nUntil the role handles it, install coreutils using the --allowerasing\noption which will remove coreutils-single at the same time. Use a\ncommand task for now since this option has just been added to\nansible:devel [1].\n\n[1] https://github.com/ansible/ansible/pull/48319\n\nChange-Id: I43bbe9dae3d6796e308fbf66cb04d16b57ff5e37\nStory: 2007612\nTask: 39607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff1a74f058cc496a74a298819a1b818beddd9e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ff1a74f058cc496a74a298819a1b818beddd9e8"}]},"branch":"refs/heads/master"},"a23d8f65e68fb933c3a33f6a6881d3c5708b1d83":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-05-12 12:28:26.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/724963/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/724963/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/724963/5"}}},"commit":{"parents":[{"commit":"576a319e71bfadda241c987c8e91ad7714098812","subject":"Blacklist Ansible 2.9.8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/576a319e71bfadda241c987c8e91ad7714098812"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 10:10:09.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-12 12:28:26.000000000","tz":0},"subject":"Install coreutils package before using configdrive role","message":"Install coreutils package before using configdrive role\n\nOn seed hypervisors running CentOS 8, the configdrive role will fail to\ninstall coreutils if coreutils-single is already present:\n\nError:\n Problem: problem with installed package coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8_1.1.x86_64\n  - conflicting requests\n\nUntil the role handles it, install coreutils using the --allowerasing\noption which will remove coreutils-single at the same time. Use a\ncommand task for now since this option has just been added to\nansible:devel [1].\n\n[1] https://github.com/ansible/ansible/pull/48319\n\nChange-Id: I43bbe9dae3d6796e308fbf66cb04d16b57ff5e37\nStory: 2007612\nTask: 39607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a23d8f65e68fb933c3a33f6a6881d3c5708b1d83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a23d8f65e68fb933c3a33f6a6881d3c5708b1d83"}]},"branch":"refs/heads/master"},"915865d43b741f69c1341a5276296c61bccd93de":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2020-05-13 16:34:47.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/724963/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/724963/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/724963/6"}}},"commit":{"parents":[{"commit":"840c56167ac104ec18f7efde7b558be77274960e","subject":"Merge \"Fix multiple CI failures\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/840c56167ac104ec18f7efde7b558be77274960e"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 10:10:09.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-13 16:34:47.000000000","tz":0},"subject":"Install coreutils package before using configdrive role","message":"Install coreutils package before using configdrive role\n\nOn seed hypervisors running CentOS 8, the configdrive role will fail to\ninstall coreutils if coreutils-single is already present:\n\nError:\n Problem: problem with installed package coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8_1.1.x86_64\n  - conflicting requests\n\nUntil the role handles it, install coreutils using the --allowerasing\noption which will remove coreutils-single at the same time. Use a\ncommand task for now since this option has just been added to\nansible:devel [1].\n\n[1] https://github.com/ansible/ansible/pull/48319\n\nChange-Id: I43bbe9dae3d6796e308fbf66cb04d16b57ff5e37\nStory: 2007612\nTask: 39607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/915865d43b741f69c1341a5276296c61bccd93de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/915865d43b741f69c1341a5276296c61bccd93de"}]},"branch":"refs/heads/master"},"fc017249c23cb819fd44202e2cff80d563cab8ee":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-05-13 20:30:08.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/63/724963/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/63/724963/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/63/724963/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/63/724963/7"}}},"commit":{"parents":[{"commit":"93c471406a60d018da1c63dc71a2e09b43bf6191","subject":"Merge \"CentOS 8: Fix network configuration persistence\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/93c471406a60d018da1c63dc71a2e09b43bf6191"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-02 10:10:09.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-13 20:30:08.000000000","tz":0},"subject":"Install coreutils package before using configdrive role","message":"Install coreutils package before using configdrive role\n\nOn seed hypervisors running CentOS 8, the configdrive role will fail to\ninstall coreutils if coreutils-single is already present:\n\nError:\n Problem: problem with installed package coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8.x86_64\n  - package coreutils-8.30-6.el8_1.1.x86_64 conflicts with coreutils-single provided by coreutils-single-8.30-6.el8_1.1.x86_64\n  - conflicting requests\n\nUntil the role handles it, install coreutils using the --allowerasing\noption which will remove coreutils-single at the same time. Use a\ncommand task for now since this option has just been added to\nansible:devel [1].\n\n[1] https://github.com/ansible/ansible/pull/48319\n\nChange-Id: I43bbe9dae3d6796e308fbf66cb04d16b57ff5e37\nStory: 2007612\nTask: 39607\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc017249c23cb819fd44202e2cff80d563cab8ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fc017249c23cb819fd44202e2cff80d563cab8ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
