)]}'
{"id":"openstack%2Fkayobe~725601","triplet_id":"openstack%2Fkayobe~master~I22a7d219c0bc4bfc2f78d2352d1a84fa555d2c69","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I22a7d219c0bc4bfc2f78d2352d1a84fa555d2c69","subject":"Constraint python-openstackclient install with requirements.txt","status":"ABANDONED","created":"2020-05-05 13:54:23.000000000","updated":"2020-05-15 20:52:52.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0eefa2b51e7b0597b4b412409e78c26f37aca752","_number":725601,"virtual_id_number":725601,"owner":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-11 14:08:47.000000000","_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-11 15:52:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"disliked":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":-1,"date":"2020-05-11 14:05:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-11 13:54:33.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":-1,"date":"2020-05-11 14:05:49.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-05-11 13:54:33.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-11 13:54:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-05-11 14:05:49.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-11 15:52:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"763293d1d41f842705bb6beb1d72044690dafe92","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-05-05 13:54:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a2c34e4a3c19f5932d74597fa801c5ec6b2a826","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-05 15:58:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/671c9685d3e341d89f28c5e043dc5c94 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6886bd4a112f4a0bb5f4e190a67b6223 : SUCCESS in 10m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1808f22d27944a6fb7e9de4813efc151 : SUCCESS in 6m 06s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d8ebd7162f1946d18a19b4542fb1c9af : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/48ae68abd2204355a0401cd626764a28 : SUCCESS in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78e3f27b2e7a4b01b41a735cf276369e : SUCCESS in 9m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a920577807294f90bf65124b6e38212b : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/badfa60374c840e3b2e1d43276de00f9 : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3fc210b15909444480589574f7ab5b26 : SUCCESS in 28m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9e27554ae4c34d2482fb14b612fffce3 : SUCCESS in 1h 17m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a9b0e59b95f04925b3b4ee67a7adb9b6 : SUCCESS in 16m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e9b675b2093c4a8abec02689740a740b : SUCCESS in 2h 02m 48s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4a2c7b6c25084736927185d967eeb3c4 : SUCCESS in 17m 39s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c5d099c6c87143f8b477fb217dda90cc : SUCCESS in 27m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ad1f94fd1dd5dd67e6e38a2cf4a84e3bd1746468","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 10:36:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5403e37d3005c604003e77b00e884366f76ae17c","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-05-11 10:39:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0c61066f573ae456842f5ae1befecf1c021649c7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 10:44:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"370e314d2fd68849bdac1127f5ecb5c6f9881df8","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-05-11 11:37:18.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"fd3d17ee6b0e93bcf08cc466f0dc2f3cf3ee2602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:33:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38a9f0c8ce1342cc87de5190f392da2c : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ac64cd486784749b1fb26479e2c72ec : SUCCESS in 7m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/700b49dbc98443febb9152c2ee738192 : SUCCESS in 7m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/841542add711427ca6caaf1cab18f822 : SUCCESS in 9m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7180b53fe23c41248e4e148f743d8e6a : SUCCESS in 6m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3be356cfd7064612a22db9be140dd112 : SUCCESS in 7m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bde2f5c2ae224d2ea2e27c94bb28948e : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5d501491e06e4f9ab4bf41f0e0a61314 : SUCCESS in 10m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d53373498174179851dca6143cc9be2 : SUCCESS in 29m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/1c5d68808119488f90f9b2d6ee256e31 : SUCCESS in 1h 06m 56s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/63a17ec6cba34b91b93c340ac121f5bf : SUCCESS in 14m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4d9f90216bd54a4f86b2f52542963f5a : SUCCESS in 1h 51m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e4638338a73341a5bbbd7d8ce9fc31f3 : SUCCESS in 18m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b8d0c7fe2075476aa4a6dc2a8fd43073 : SUCCESS in 27m 35s","accounts_in_message":[],"_revision_number":2},{"id":"38bb11664625d0ae231e46220d78f4837e9ac994","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 13:39:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9927e86d0ded8ccd7e87789e02e4dcd3bf8afe04","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-11 13:54:33.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99ea5a65122d9207a78f6cd4a7016e95e75ff3b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 13:54:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"83ceb3d6d93e7dc2b333d762da56c5c89cc23851","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 14:05:49.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"7d78cd3dd53f3f0acb7d709fd90bc3256e64d0ee","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 14:05:54.000000000","message":"Patch Set 2: Code-Review-1","accounts_in_message":[],"_revision_number":2},{"id":"546aefde68045905edead506b20e7a85afa6ca09","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 14:06:33.000000000","message":"Patch Set 2:\n\nJust realised we don\u0027t need this on master, since we bumped cliff to 3.1.0. I\u0027ll cherry pick it.","accounts_in_message":[],"_revision_number":2},{"id":"8548e1fdde04be8f4ca7d7aaf779486313bd6a05","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 14:06:41.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit eba57aff5b9ab3e4fda81ca783795f6caeed4c34","accounts_in_message":[],"_revision_number":2},{"id":"9565fcf8f7ba20f5a4ddbccca43e2150fe5c72ca","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 14:06:49.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit ff8e9be89e06d25da3feb51bc3f9acd0f2bc06a5","accounts_in_message":[],"_revision_number":2},{"id":"1708193099a8f9c85bf6b7c5352e057d35719164","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 14:06:57.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/rocky as commit b49aada3d836421129d9f9b5899347a86e3bad78","accounts_in_message":[],"_revision_number":2},{"id":"560b7e71c91f1f92e4b3b9742eed2714074a2c2f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-11 14:07:43.000000000","message":"Patch Set 2:\n\n\u003e Just realised we don\u0027t need this on master, since we bumped cliff\n \u003e to 3.1.0. I\u0027ll cherry pick it.\n\nI thought we would merge it for cherry-picking it, and keep it to have the installation method the same across all branches.","accounts_in_message":[],"_revision_number":2},{"id":"7657d3ce017794795f89ae63fd247d5116117b9a","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-05-11 14:08:47.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nIt is still a reasonable practice to constraint using kayobe requirements in case something else breaks in the future.","accounts_in_message":[],"_revision_number":2},{"id":"c86c50f31a5b7d7eb9a406629708dc3267dc2f77","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 15:52:31.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da9de6b0a7f844e5bec2e301d137bf22 : SUCCESS in 7m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d22df053bb34e83817b8acb43959f6e : SUCCESS in 6m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9bfdbb2652884eb8b0f0f4d756ff2fbb : SUCCESS in 7m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d6707ad2e5c45e38f5e826b7aff49a7 : SUCCESS in 7m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c487c623ff9a48ab9b14163733758729 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/153539a6c2534936bef5a8c7b7d29de1 : SUCCESS in 11m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/317c5a4e7c1b44b59a3f9f1599817616 : SUCCESS in 34m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/512799b721214670bb34bcb68975176c : SUCCESS in 1h 09m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/eb403b1700e543119f991f563823b6ee : SUCCESS in 18m 31s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e8ab272084cb4e1492e7423b153bdd1a : SUCCESS in 1h 52m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ca9462fe66004458aee8567bdc353ada : SUCCESS in 21m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/0390cd308139479e800e2bb823b573be : SUCCESS in 27m 58s","accounts_in_message":[],"_revision_number":2},{"id":"85e0d87901a514c1e879bd7ecd1c84731174912f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-12 09:30:59.000000000","message":"Patch Set 1:\n\n\u003e (1 comment)\n \u003e \n \u003e It is still a reasonable practice to constraint using kayobe\n \u003e requirements in case something else breaks in the future.\n\nIt just seems like a bit of a fudge to me. Maybe we should use a separate virtualenv for the clients?","accounts_in_message":[],"_revision_number":1},{"id":"456928734acad53679bac5633c73606ef5c08c5d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-12 09:43:45.000000000","message":"Patch Set 2:\n\n\u003e \u003e (1 comment)\n \u003e \u003e\n \u003e \u003e It is still a reasonable practice to constraint using kayobe\n \u003e \u003e requirements in case something else breaks in the future.\n \u003e \n \u003e It just seems like a bit of a fudge to me. Maybe we should use a\n \u003e separate virtualenv for the clients?\n\nSure, like we do in AUFN init-runonce.","accounts_in_message":[],"_revision_number":2},{"id":"0eefa2b51e7b0597b4b412409e78c26f37aca752","author":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"date":"2020-05-15 20:52:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"30e9b4cdf48a0b1e1445b3da2fa08c10804bf926","revisions":{"34b7196fef00717d93293696e643ded92e090088":{"kind":"REWORK","_number":1,"created":"2020-05-05 13:54:23.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/01/725601/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/725601/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/725601/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/725601/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/725601/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/725601/1"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-05-05 13:50:34.000000000","tz":60},"committer":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-05-05 13:54:19.000000000","tz":60},"subject":"Constraint python-openstackclient install with requirements.txt","message":"Constraint python-openstackclient install with requirements.txt\n\nWithout this constraint, we get this error when attempting to run kayobe\ncommand under kayobe-venv after overcloud-test-vm is deployed because\nunsupported version of cliff gets installed:\n\n    invalid conflict_resolution value: \u0027ignore\u0027\n\nChange-Id: I22a7d219c0bc4bfc2f78d2352d1a84fa555d2c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34b7196fef00717d93293696e643ded92e090088"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34b7196fef00717d93293696e643ded92e090088"}]},"branch":"refs/heads/master"},"30e9b4cdf48a0b1e1445b3da2fa08c10804bf926":{"kind":"REWORK","_number":2,"created":"2020-05-11 11:37:18.000000000","uploader":{"_account_id":28022,"name":"Bharat Kunwar","email":"brtknr@bath.edu","username":"brtknr"},"ref":"refs/changes/01/725601/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/725601/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/725601/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/725601/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/725601/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/725601/2"}}},"commit":{"parents":[{"commit":"565a0614dc322dfe38a62b3fe883137ebe64773c","subject":"Merge \"Define default variables for Monasca config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/565a0614dc322dfe38a62b3fe883137ebe64773c"}]}],"author":{"name":"Bharat Kunwar","email":"brtknr@bath.edu","date":"2020-05-05 13:50:34.000000000","tz":60},"committer":{"name":"Bharat Kunwar","email":"bharat@stackhpc.com","date":"2020-05-11 11:37:11.000000000","tz":0},"subject":"Constraint python-openstackclient install with requirements.txt","message":"Constraint python-openstackclient install with requirements.txt\n\nWithout this constraint, we get this error when attempting to run kayobe\ncommand under kayobe-venv after overcloud-test-vm is deployed because\nunsupported version of cliff gets installed:\n\n    invalid conflict_resolution value: \u0027ignore\u0027\n\nChange-Id: I22a7d219c0bc4bfc2f78d2352d1a84fa555d2c69\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/30e9b4cdf48a0b1e1445b3da2fa08c10804bf926"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/30e9b4cdf48a0b1e1445b3da2fa08c10804bf926"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
