)]}'
{"id":"openstack%2Fkayobe~726853","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I68fb65cd3bc7c4390813747f9ac4ecebc3b4145f","project":"openstack/kayobe","branch":"stable/train","topic":"story/2007654-stable/train","hashtags":[],"change_id":"I68fb65cd3bc7c4390813747f9ac4ecebc3b4145f","subject":"Remove newline from default public_net_name","status":"MERGED","created":"2020-05-11 13:59:51.000000000","updated":"2020-06-19 09:31:38.000000000","submitted":"2020-06-19 09:28:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"726853-1592558913437-93a3f69b","meta_rev_id":"990f1e441cc38480defbc57cb7c83a4946878742","_number":726853,"virtual_id_number":726853,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 09:28:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-18 21:21:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-19 07:23:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 07:23:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 07:23:34.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-19 09:28:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d1952fe641ef124dbe65823defc0926b4eb528b3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 13:59:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd18b1250fc82e101f8e5e1ec1456682d6e89e31","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-11 14:03:01.000000000","message":"Patch Set 1: Workflow-1\n\nWrong branch","accounts_in_message":[],"_revision_number":1},{"id":"747373386a8d6174d4cb43c289acfea440742b4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-11 16:28:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e01333af58e94401a0000526e325ac56 : SUCCESS in 9m 20s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/92df16f2e3ba406fae553cf0e96d20d3 : SUCCESS in 6m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f28e501c69ab47aba9ce630c7d1a3806 : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d28c4ff903046bab08a102bfeea6600 : FAILURE in 11m 19s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c9f4418da24d42058bdf851597957684 : FAILURE in 11m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36e0371db3ff4686bcf8e30564d86a7e : SUCCESS in 8m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/81099928842e4710bb49ffd3176c0e4d : SUCCESS in 6m 22s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0deb46f23fff4ea7bc2779f808473ab8 : SUCCESS in 8m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e15c148bef73420e8c31ba573a244280 : SUCCESS in 12m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7a97a55fe60a4f5b91bada62ad2014b0 : SUCCESS in 32m 25s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/da9e176d9b4648529d468b92385ce077 : RETRY_LIMIT in 8m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a8119bdc4a2a4884bdab69c8e53a00f4 : SUCCESS in 1h 23m 35s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/14e8a5e4c7ca47b49fbd19fd3544932e : SUCCESS in 1h 55m 19s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/fb649e97eb174ee794da904f2b9a91b3 : RETRY_LIMIT in 4m 46s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2b7e2d7f62484b56b37e54e4f9d947a9 : RETRY_LIMIT in 5m 38s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8946f5f064584ebaac1e793719453bc3 : RETRY_LIMIT in 5m 18s","accounts_in_message":[],"_revision_number":1},{"id":"8378da6a2a54f9e15f528d5b98c9dc04925eee7f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-05 09:55:29.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c8f3d9f9e39389841642f036e6fee30cde319c38","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-05 11:05:08.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c41c80da5eed16f3373b2acf7f353c5e46a6429c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 12:10:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4932e921765649b48b8f402e98aef5af : SUCCESS in 7m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/008f02536d8f4f84b2da023c0a5e1979 : SUCCESS in 4m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/41f4d5d6151b47f79860e3b6008f8892 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6b04e94bc00435794975d08ab080f28 : SUCCESS in 4m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/8242d591bcd64ebd8b7b1929d7143d7d : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3237e8b24b234ad6a6345030f39970e2 : SUCCESS in 7m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0761d5c58e94c1db624d13b2813799a : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/198752e3c0c848e2884e226463f02290 : SUCCESS in 4m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a59050ed4e524d0aa3abe86afc707889 : SUCCESS in 7m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c6c701702554672890076e7a95dacba : SUCCESS in 34m 28s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e3b47944c293448d9554878238636940 : SUCCESS in 1h 10m 41s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/06235b2d1d99464095ccd87f9b6f4832 : SUCCESS in 59m 03s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/8ab733442f6f4d35a30e102eabc256fb : SUCCESS in 13m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5d2075384d5241a4bcd63ad5b052baed : SUCCESS in 16m 29s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3bd587bbdf554a84a0890fb3dcb40603 : SUCCESS in 2h 04m 09s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/c1f07d72254b48c58d82d616076ac9fa : SUCCESS in 21m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/baa5f42746344018914363cca2f2c706 : SUCCESS in 19m 19s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e4857eded9714530a2fa231d91357f36 : SUCCESS in 31m 44s","accounts_in_message":[],"_revision_number":2},{"id":"b45a0c1509ad95f17fb8f60df96303b2955c52d7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 10:04:30.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"707f19e0e225f5ba1194435fa340be56453000e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 10:04:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"68258d0d59b6f3ee427e53288461853cde369e5a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 11:26:18.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7778c7342b2e4dfa9f46e5908be13960 : SUCCESS in 7m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/26c75b6f441043ea8d7a712849f2b4ee : SUCCESS in 9m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/841f7dd79c2c4521a10ce77ba1c0d6af : SUCCESS in 8m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/a09a9524b47b41d4a4a2c0001342353f : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f05d2cba5c5d497180eb2ec8681709de : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5e80d6958d343bab9213124bef89778 : SUCCESS in 5m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5beddc09943948bc865a43cdcc353863 : SUCCESS in 7m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c0caf8f756304343b3bcecd6caaee52f : SUCCESS in 10m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e53e9d2b0cf74653a2da40b15350fff5 : SUCCESS in 38m 53s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/66d8aa2d42de4493a4dda2a005550ef4 : FAILURE in 1h 02m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8ea65208a88a4cb3b5d9eb31f84fc6e3 : SUCCESS in 1h 15m 17s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/1a93262db47f4ef2a77caba47fa6e650 : SUCCESS in 17m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0bbdced0f41b4354b69a77d0ba32044c : SUCCESS in 21m 54s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/03d61a8cb6af40ed971c2a7622d760b1 : FAILURE in 1h 04m 18s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/ae423ead12b441e4af7127741b970f12 : SUCCESS in 23m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/07f17cd2cdd74c819f07398403c1e908 : SUCCESS in 21m 07s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/744e8bb1e0354b02b32b09950d4d5475 : FAILURE in 29m 17s","accounts_in_message":[],"_revision_number":2},{"id":"9d05aefe3f53444bbd826e27400025fbbbced80c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 21:21:18.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"2f6deaae6bc066c6078e6b25f0ec100d76b9d988","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 23:35:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54f8d4fbf3544214b2320f3c970cf42f : SUCCESS in 9m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a4092940f6d04184aa2b4d4c4aa7b16f : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/953e81a735d24f1c8daafbe5cbc8af39 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3ab1f5704a342e0874ce19d46adda95 : SUCCESS in 8m 24s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/87a652cdd6bf471b93f2d1895596037d : SUCCESS in 5m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d17f563e2c74950a13a6d1bd075be22 : SUCCESS in 7m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57cdb229c7c547078303cb711247bc4c : SUCCESS in 5m 55s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/61342285dd3e403f8944edbcfc46ab11 : SUCCESS in 7m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6a5b6e4951844a3e96b74447725e6fb0 : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c1853477aaf1456ab79ad99ebd9a4c76 : SUCCESS in 35m 46s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/f067aa507d1c446e8f87e606e23d48c4 : SUCCESS in 1h 10m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9ce548e624b4467f93f95f7f882dc58f : SUCCESS in 1h 14m 46s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/6964e23527f144969411176908d76114 : SUCCESS in 33m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e920c287307a4de1a36dadda98814432 : SUCCESS in 20m 06s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3ca8d58546c54b5bb3d6256d14400430 : SUCCESS in 2h 13m 03s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/07d24336c747459d84ff5a162e10868c : SUCCESS in 24m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e26c0d309e8b41128ef86f65742b5508 : SUCCESS in 22m 37s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/910938695b8d435cb449673acec0c574 : SUCCESS in 37m 57s","accounts_in_message":[],"_revision_number":3},{"id":"3a679e98f220c6e78093737971a1853cfdeffa39","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 07:23:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"502ba5f71e83b8714c79552b33dde317b25b9336","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 07:23:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"b7814db4fa499e7de135cf5c9d2ad6030983fc26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:28:33.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/af67ecd52fca494b800d8d9ab6ac22b4 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b64274cc56744994aea636723e2f4de1 : SUCCESS in 7m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/21c33b81a24e483baebca3bb1ec185a3 : SUCCESS in 6m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e38548b1c47b4cfdb64c42f8b0ecacc3 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a973839c554ff4aab9dedfc3fd4f50 : SUCCESS in 14m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/baa8039a33c64185bc71f0b315f2c331 : SUCCESS in 6m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c63fcb9226af4505ad179b082e2251f3 : SUCCESS in 7m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/54b95e9ed95247d1b364a991939738f3 : SUCCESS in 14m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3653ea3a29d147c79fa78f52c0e95f24 : SUCCESS in 39m 49s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/66d36636d36940e8be2148c8e5674b09 : SUCCESS in 1h 17m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e62f36f1336f4b6685c4b8bc9d8d5d37 : SUCCESS in 1h 10m 01s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/4fa01dc9f48e4ba49837154fb1154557 : SUCCESS in 18m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/cdf56ccde35a454ca58033220ca8a53a : SUCCESS in 20m 17s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f7a6702b72f8462eb95b011df13ec2f8 : SUCCESS in 1h 59m 54s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/2a389f5854904d548b196729b54413a2 : SUCCESS in 22m 33s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/9c61ab05e9ee4cb3b48c5cd459f63c4a : SUCCESS in 19m 27s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3ffdccd4b5cf428e87fc40fe7b08e1f1 : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":3},{"id":"143319fdd694b78456fe6efa19ed2bb6d2de2d95","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:28:33.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"711aa839128f0b764929344453b4866e717dce5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:31:38.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60381b5232fa40db889ead20d8970ea3 : SUCCESS in 2m 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ebf38608e3147888a868857c5a13e8d : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"49cef61f50b64220e4ede1cac95c582d837fe2c2","revisions":{"2f273fc5775e2c2071c7aa2b2661051fb87b7f04":{"kind":"REWORK","_number":1,"created":"2020-05-11 13:59:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/53/726853/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/726853/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/726853/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/726853/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/726853/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/726853/1"}}},"commit":{"parents":[{"commit":"683ae96b320d7a5a3f2ba78effd640e0f805c33e","subject":"Merge \"Fix loading of Monasca Grafana dashboards\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/683ae96b320d7a5a3f2ba78effd640e0f805c33e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-11 13:52:25.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-11 13:59:35.000000000","tz":60},"subject":"Remove newline from default public_net_name","message":"Remove newline from default public_net_name\n\nThe public network name is set via public_net_name, and its default\nvalue is the first item in the external_net_names list. The variable\nuses the YAML folded block scalar notation, but does not use the\nchomping indictator, which means the string is assigned a trailing\nnewline.\n\nThis change fixes the issue by adding the chomping indicator.\n\nChange-Id: I68fb65cd3bc7c4390813747f9ac4ecebc3b4145f\nStory: 2007654\nTask: 39731\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f273fc5775e2c2071c7aa2b2661051fb87b7f04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2f273fc5775e2c2071c7aa2b2661051fb87b7f04"}]},"branch":"refs/heads/stable/train"},"e83a8cef15ebb998c74aa847be00c311e68fe272":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2020-06-05 09:55:29.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/726853/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/726853/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/726853/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/726853/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/726853/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/726853/2"}}},"commit":{"parents":[{"commit":"d932d4357fe3712ae90aedba2ee495629e66159f","subject":"Fix docker-devicemapper playbook","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d932d4357fe3712ae90aedba2ee495629e66159f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-11 13:52:25.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-05 09:55:11.000000000","tz":120},"subject":"Remove newline from default public_net_name","message":"Remove newline from default public_net_name\n\nThe public network name is set via public_net_name, and its default\nvalue is the first item in the external_net_names list. The variable\nuses the YAML folded block scalar notation, but does not use the\nchomping indictator, which means the string is assigned a trailing\nnewline.\n\nThis change fixes the issue by adding the chomping indicator.\n\nChange-Id: I68fb65cd3bc7c4390813747f9ac4ecebc3b4145f\nStory: 2007654\nTask: 39731\n(cherry picked from commit 6beaf5b5e38c3a90651afb67899d8ee6661e2cf9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e83a8cef15ebb998c74aa847be00c311e68fe272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e83a8cef15ebb998c74aa847be00c311e68fe272"}]},"branch":"refs/heads/stable/train"},"49cef61f50b64220e4ede1cac95c582d837fe2c2":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-18 21:21:18.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/53/726853/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/53/726853/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/726853/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/726853/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/53/726853/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/53/726853/3"}}},"commit":{"parents":[{"commit":"8039e6d4742401c56bbde3338f3a04ea445cba4f","subject":"Merge \"Fix ironic inspector store endpoint configuration\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8039e6d4742401c56bbde3338f3a04ea445cba4f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-11 13:52:25.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-18 21:21:18.000000000","tz":0},"subject":"Remove newline from default public_net_name","message":"Remove newline from default public_net_name\n\nThe public network name is set via public_net_name, and its default\nvalue is the first item in the external_net_names list. The variable\nuses the YAML folded block scalar notation, but does not use the\nchomping indictator, which means the string is assigned a trailing\nnewline.\n\nThis change fixes the issue by adding the chomping indicator.\n\nChange-Id: I68fb65cd3bc7c4390813747f9ac4ecebc3b4145f\nStory: 2007654\nTask: 39731\n(cherry picked from commit 6beaf5b5e38c3a90651afb67899d8ee6661e2cf9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49cef61f50b64220e4ede1cac95c582d837fe2c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49cef61f50b64220e4ede1cac95c582d837fe2c2"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
