)]}'
{"id":"openstack%2Fkayobe~727686","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I9eee17e99af008252ec226dca25c38bb3b0b7855","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"I9eee17e99af008252ec226dca25c38bb3b0b7855","subject":"Use ensure-docker role instead of install-docker","status":"MERGED","created":"2020-05-13 09:38:06.000000000","updated":"2020-05-13 20:47:02.000000000","submitted":"2020-05-13 20:45:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"727686-1589402712961-d71d2c75","meta_rev_id":"0b35ed872acc804bf98d192b342aabafbfd19755","_number":727686,"virtual_id_number":727686,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-13 20:45:12.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-05-13 18:45:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-05-13 18:45:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-13 18:45:44.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-13 20:45:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e0dde05525f7a6933be6f32364b52f5d05d483a7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-13 09:38:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bed088f07a414986ffdd567c37dad18ce963f9a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 11:41:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/656a3281e9674c6298c7f91cb505b490 : SUCCESS in 6m 10s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c038847f1c574f828c3825d2743d9e8a : SUCCESS in 5m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e3b5be4eecac403dbb3cedfc26c166c6 : SUCCESS in 4m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/591f26aebfee410ea49fc4e6cd46dc96 : SUCCESS in 4m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/19f6635a610e49d0a4aa5dec006700ce : SUCCESS in 6m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecdf66dd55884bd7b1e5767c92b6dda3 : SUCCESS in 6m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/41e9313d811a4423811dbc48a25c4696 : SUCCESS in 5m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6693f98f42c8403585c56568cf90d153 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/54714048391646afab8da0d12e0c7e3b : SUCCESS in 36m 02s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/56aa771f2e9e459f83d6928ab7b27aee : SUCCESS in 1h 10m 15s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/54cd8fd193654da6961422e06a510791 : SUCCESS in 1h 56m 49s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/7346dabcd96f48b1bdbd8e4883afa520 : SUCCESS in 19m 37s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d82ca15e77cb4902b9db850e1dad6bdb : SUCCESS in 31m 31s","accounts_in_message":[],"_revision_number":1},{"id":"57be1660ac2862f83fa8bad8ca1dad2b67a7e3fe","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-13 18:45:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4c90d2a65be7b81496d2b378f4a44d52ce1172e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 18:45:57.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"292874d78927f55af09f6787e545cb921433b9d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 20:45:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a44bd5a68a534b679e7ab3189e1537bd : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3336daec3374dc4b62355ff7173952d : SUCCESS in 7m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00017e4e65ed44878de30a14da99482a : SUCCESS in 6m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c0fa9a74f0174194b515a53fb89530f2 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a708c19d5834ddfacb02b4e368450c3 : SUCCESS in 4m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f3771da703d0476ba14c91bfc3231f56 : SUCCESS in 9m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/891dfc51dc6e4a9b9e1a363c1e801e54 : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6055f61679c84b5a9bc796eca2d2a6d2 : SUCCESS in 33m 56s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/84dfb1c9e1664b19999e2dd9502e89bc : SUCCESS in 58m 55s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ddef05e0504a4f819d6382bb177e2247 : SUCCESS in 1h 53m 12s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/91406590b868474a9522ac4b53a47e5e : SUCCESS in 28m 15s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/39ef2e796cc24f87be4839bc80f3bd11 : SUCCESS in 41m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a339117d929648f67a4d6845a8907f1535df2e84","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 20:45:12.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"0b35ed872acc804bf98d192b342aabafbfd19755","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-13 20:47:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3682c988b3064ed38eac7c70c52292a3 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"53497b626bf50f53b71587e60af5b5a0561caec4","revisions":{"53497b626bf50f53b71587e60af5b5a0561caec4":{"kind":"REWORK","_number":1,"created":"2020-05-13 09:38:06.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/86/727686/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/86/727686/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/727686/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/727686/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/86/727686/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/86/727686/1"}}},"commit":{"parents":[{"commit":"cbf94c97333be76de72dde40e2f8d791d8df8dee","subject":"Merge \"Docs: Configure firewall to allow testing of baremetal\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/cbf94c97333be76de72dde40e2f8d791d8df8dee"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-13 09:37:11.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-13 09:37:59.000000000","tz":120},"subject":"Use ensure-docker role instead of install-docker","message":"Use ensure-docker role instead of install-docker\n\nThe install-* roles are being deprecated and renamed to follow the\nensure-* naming convention [1].\n\nThis is extracted from change Iab1d84e6a8c1b3dd81e53279309153687677a061\nin the master branch.\n\n[1] http://lists.zuul-ci.org/pipermail/zuul-announce/2020-April/000071.html\n\nChange-Id: I9eee17e99af008252ec226dca25c38bb3b0b7855\n(cherry picked from commit 0d699b53f4a9086cada86d51e53f00594ec94cd1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53497b626bf50f53b71587e60af5b5a0561caec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53497b626bf50f53b71587e60af5b5a0561caec4"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
