)]}'
{"id":"openstack%2Fkayobe~729622","triplet_id":"openstack%2Fkayobe~master~Ic9db7f84e9f0dcbe69c31b6fff6bda3dfb401a33","project":"openstack/kayobe","branch":"master","topic":"story/2004959","hashtags":[],"change_id":"Ic9db7f84e9f0dcbe69c31b6fff6bda3dfb401a33","subject":"CI: Encrypt passwords.yml in overcloud host configure job","status":"MERGED","created":"2020-05-20 14:38:32.000000000","updated":"2020-05-28 18:06:38.000000000","submitted":"2020-05-28 17:51:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729622-1590688318369-69ea3769","meta_rev_id":"2ce285fc6a6987f7e8d79e3e5fdc49e93243a97e","_number":729622,"virtual_id_number":729622,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2020-05-28 15:43:39.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 17:51:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-05-28 15:43:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-28 14:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-05-28 15:43:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 14:17:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-05-28 15:43:55.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-05-28 17:51:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f6c93a261a9223341b35a66c5f5b59d20acac42c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-20 14:38:32.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"821258cca2a09a6b3728085f335253986da8ab90","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-20 16:48:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/65b088de5c79490eac4d03362e94d407 : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7eef96db1de40639c4c7d2503f71266 : FAILURE in 5m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc02e0f9880d4653b7f9bfb2fef9de2b : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/fe5a0fb10282479782823f83493a12a1 : SUCCESS in 6m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45e4e3e39b04447eaba2d113c47c183e : SUCCESS in 5m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ba65a4ccc424a3eaf35f047046b497f : SUCCESS in 8m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ac50211e65b246f2a813e05721081aa5 : SUCCESS in 6m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/788a35f874014523a3377e961ebcdcc5 : SUCCESS in 10m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/83e04edb65e9449585d50cf78d1cc80f : SUCCESS in 34m 34s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c0c7195080e24e33880b03061b9080b5 : SUCCESS in 1h 07m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/f1b7988040ec405a8e3c7c0d27b2c1fc : RETRY_LIMIT in 5m 56s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9d0e08ee5c044c908e16406b51e00e10 : SUCCESS in 1h 55m 58s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/267fb33bfa42487f9ea7c13157b84340 : SUCCESS in 17m 44s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f2e51a9e2a8c4c28878ec3c05c50be95 : SUCCESS in 34m 35s\n\nWarning:\n  Comments left for invalid file kayobe/tests/unit/test_vault.py","accounts_in_message":[],"_revision_number":1},{"id":"4c118953da26c12efade96d1c0cb786037b0ed56","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 07:53:36.000000000","message":"Topic changed from copy-certificates to story/2004959","accounts_in_message":[],"_revision_number":1},{"id":"607086449f673b920502cbff4262b8c075b97638","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 07:55:04.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"0af2509bdbc58e8f1b74b805b9638e0f09acb17c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 07:55:12.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"fd74ff4814b96a07392ab4def6adbdbe7c9c4278","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 08:38:07.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/58733c9e9e1d45ccbb50ffbf9c3f9062 : RETRY_LIMIT in 5m 26s","accounts_in_message":[],"_revision_number":3},{"id":"22a510c21f81b312e92b4fcbb47bb1c7bf1a51da","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 08:54:57.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a52840c9f0692f16d8091a14836e859c0beb19e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 09:35:44.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a80e94e3d6c34e9c964c97c9d85a2e7c : RETRY_LIMIT in 5m 19s","accounts_in_message":[],"_revision_number":4},{"id":"5514d2d74335587e74f636d7728e5ecb1241f875","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 09:42:44.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"f9922c26f51bf25517dd8ec3050b99a673e73b8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 10:05:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1611c7bc344e4fc987f185f1d6d0f128 : SUCCESS in 15m 38s","accounts_in_message":[],"_revision_number":5},{"id":"f394472ac6ca5be97b026209e727ee8455a9911b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 10:13:36.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"2a79e08ab8159d42e845e74aeccdf9c879e588b6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 10:13:45.000000000","message":"Patch Set 6:\n\nReady for review","accounts_in_message":[],"_revision_number":6},{"id":"4dd4c217d0811646e2f59a6710fbe7429c1f75b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 12:03:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b8180f03c19d40afa4b298ea8a9f3e46 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c74c0bbe00d5404287f455cceff92534 : SUCCESS in 4m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0034fac670964ab1b8a553bdd576860a : SUCCESS in 4m 41s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f6a305db8833495187fce3664f606944 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d34be11f5c0d413886f791a7cead7d13 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1786e4596b2649aea4eec38b2c908251 : SUCCESS in 12m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ef6fb5dc4fe14daeafca78a9b34cfd46 : SUCCESS in 4m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fbf9b52b427b434cbd4916d02c10a1f7 : SUCCESS in 8m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1b27678d6cb5433796a283719d2728b5 : SUCCESS in 27m 13s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d9e4af1fecc54f409a91d6d9eeada60f : SUCCESS in 1h 05m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/62c0b8c1d977457aa368beb7dc78260a : SUCCESS in 15m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7fffc056a0574bc08c04d45311e334e6 : SUCCESS in 1h 48m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/322c32c5b19e4bb788a5ae1533b4608f : SUCCESS in 19m 49s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2f776e679f6c45acbfc64e4dc8e7f1c0 : SUCCESS in 27m 11s","accounts_in_message":[],"_revision_number":6},{"id":"546edb363a4b8441dff21537ac12a73571c6889e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-27 08:38:01.000000000","message":"Patch Set 6:\n\nThere\u0027s no way the private SSH key displayed in logs could be used to access anything, right?","accounts_in_message":[],"_revision_number":6},{"id":"cb866fd074db03de59d587c8aa318493d2203cf9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 09:59:37.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"2b818b585e4a24cd0632b9bf8a3cf1aae230a099","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 12:01:50.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3fcac24d3c714e0c8f9fc0350f8e7e63 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de78e8ad674b40d09052f4c867961a35 : SUCCESS in 4m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b07a51e03d8948658e0642438468abd1 : SUCCESS in 5m 04s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f19da1335266480897fd48c5a6051dd6 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6aac50d578b44463a8853401fcbb2688 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e6f9d9954144738a327f2399515c44e : SUCCESS in 6m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6f4fe4184c58409583978ac90424d461 : SUCCESS in 6m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/306fb03583bc45c690015bbb3d6e1d9d : SUCCESS in 10m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5a8f7401dc914e8fabcce42dc4104707 : SUCCESS in 34m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6cdde216d3634eff874fe7c53e2203a7 : SUCCESS in 1h 12m 46s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ec2987054fe744dd9176296b2e57db6a : SUCCESS in 18m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e7c75be5b61343699e7eef00ae60ef64 : SUCCESS in 1h 55m 27s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c5faeae826eb49ec97f6f7d3db113b2e : SUCCESS in 20m 50s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2a0a6907bfeb4dd2b313462f1d34bc53 : FAILURE in 31m 56s","accounts_in_message":[],"_revision_number":7},{"id":"dfc268149dda9465ca4fdfcc24751046023e664e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 12:10:30.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"ac4fb48984d6d33c760c0330f141b81f0e4f60bb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 12:11:00.000000000","message":"Patch Set 8:\n\n\u003e There\u0027s no way the private SSH key displayed in logs could be used\n \u003e to access anything, right?\n\nI\u0027m not certain, so switched to a decrypt instead of view.","accounts_in_message":[],"_revision_number":8},{"id":"eecd2c1ebeb20aad15a0f27362b86178f64f4f9c","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-28 14:17:02.000000000","message":"Patch Set 8: Code-Review+2\n\nLGTM assuming it passes CI again.","accounts_in_message":[],"_revision_number":8},{"id":"b52195c2d1b848dd0d29df0ad2ab2fa513472659","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 14:23:20.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/904df37d6ebb44b0b938d4f1304301ef : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f23badc9566d484384f774982c7e71d8 : SUCCESS in 6m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/af82113482ac4aaca9395451a329bd6b : SUCCESS in 4m 30s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/9c2a3fcb1a684ea584d690628eaba026 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a0be3c4eb404c44a7c8e1f2be2f902c : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5571e26ebc24e63a860d51b8ad9f14c : SUCCESS in 8m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d27936a88dad4a45a9635ee8c6bb069c : SUCCESS in 5m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/13698627a9db496fbf3a343831dfdebb : SUCCESS in 10m 21s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4818b5314226408d80e43807cdfe0636 : SUCCESS in 33m 42s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/dd06dd94347046cfb12a1bd53c749a20 : SUCCESS in 1h 11m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/76eead7c0f1249cf87fa69ed22836c6f : SUCCESS in 18m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f403e89b7cb64bd19f1997ec3733e628 : SUCCESS in 2h 04m 50s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fdf820c224504bd8affec1107e7c35b6 : SUCCESS in 17m 53s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/307c66bebe064b5ba37656a8cb76c156 : SUCCESS in 28m 19s","accounts_in_message":[],"_revision_number":8},{"id":"a30363bfe409746fad248f2cf74837997778b712","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-28 15:38:48.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"6658648f44f372b5bc268bf2dc0f060f943535ab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:39:00.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"46b89e0031ec6dbc4684d4ef22ed58bb1e42a725","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-28 15:42:48.000000000","message":"Patch Set 8: -Code-Review -Workflow\n\nHmm, I don\u0027t see any output in the console tab. Shouldn\u0027t we see the passwords there?","accounts_in_message":[],"_revision_number":8},{"id":"2073f16da50307f6157c02a6efd88062b471a92a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 15:43:39.000000000","message":"Patch Set 8:\n\n\u003e Hmm, I don\u0027t see any output in the console tab. Shouldn\u0027t we see\n \u003e the passwords there?\n\nI switched from view to decrypt on Pierre\u0027s suggestion to avoid leaking private SSH keys.","accounts_in_message":[],"_revision_number":8},{"id":"693483e4e0bf92eeceb5cc86b809be0af354aab9","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-05-28 15:43:55.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\nIgnore that - using decrypt and view","accounts_in_message":[],"_revision_number":8},{"id":"30c7ef9c5426adc0edcfaad465ff0f9fbb46bb3c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 17:51:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":8},{"id":"a77fa36d93db848be4f224aa0ca6645758862a49","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 17:51:58.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7cb8f3df5b546d5aba23cdb6af82fad : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/34ccd5e8bf41482fbb391dfc35b805e8 : SUCCESS in 5m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/271485e11b7b4f37aa924a517a540900 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91f514175dc44081a761b8636dcabc23 : SUCCESS in 7m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/565850f387f54c47ad4c8ea76830f15f : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6dedeb881fa646b980a1df761f05a6a3 : SUCCESS in 10m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a1fb6fc4c3394ccb964d9a0329987221 : SUCCESS in 27m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/70662362989e46308db24c636935544a : SUCCESS in 1h 09m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2bf0b65ffb964af496975893c9984677 : SUCCESS in 16m 51s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b142d0c086034cbeb714a1cc5f3304ec : SUCCESS in 2h 07m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0e1e49844af045d4b4d33262afc7e92b : SUCCESS in 18m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9f0c595410c74ccbafb0a3cb0392889b : SUCCESS in 31m 23s","accounts_in_message":[],"_revision_number":8},{"id":"5c0ec05ca70ebc66ff076a294d197aa99790f81c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 17:53:38.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c6e3bf26c384e46a8482f75de2a2afa : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":8},{"id":"2ce285fc6a6987f7e8d79e3e5fdc49e93243a97e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 18:06:38.000000000","message":"Patch Set 8: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 64118fabcceb525f070978ad3e6b4393dbc228d9","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"3b03bfe3fd52a7d2ef3231629eb96957ec551295","revisions":{"c56fe5b3b73e5932d7b8c600f39f5aa8490201ae":{"kind":"REWORK","_number":1,"created":"2020-05-20 14:38:32.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/729622/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/729622/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/729622/1"}}},"commit":{"parents":[{"commit":"f99fdca17f0ba5ceda2bf002a6c9f0040342e25b","subject":"Python 3: fix Ansible Vault password helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f99fdca17f0ba5ceda2bf002a6c9f0040342e25b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 14:37:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 14:37:22.000000000","tz":60},"subject":"CI: Encrypt passwords.yml in overcloud host configure job","message":"CI: Encrypt passwords.yml in overcloud host configure job\n\nThis gives us some test coverage with Ansible Vault.\n\nChange-Id: Ic9db7f84e9f0dcbe69c31b6fff6bda3dfb401a33\nStory: 2004959\nTask: 39813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c56fe5b3b73e5932d7b8c600f39f5aa8490201ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c56fe5b3b73e5932d7b8c600f39f5aa8490201ae"}]},"branch":"refs/heads/master"},"e740daf074a50d797293771ce45b1d2d35f9246e":{"kind":"REWORK","_number":2,"created":"2020-05-21 07:55:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/729622/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/729622/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/729622/2"}}},"commit":{"parents":[{"commit":"f99fdca17f0ba5ceda2bf002a6c9f0040342e25b","subject":"Python 3: fix Ansible Vault password helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f99fdca17f0ba5ceda2bf002a6c9f0040342e25b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 14:37:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 07:55:01.000000000","tz":0},"subject":"CI: Encrypt passwords.yml in overcloud host configure job","message":"CI: Encrypt passwords.yml in overcloud host configure job\n\nThis gives us some test coverage with Ansible Vault.\n\nChange-Id: Ic9db7f84e9f0dcbe69c31b6fff6bda3dfb401a33\nStory: 2004959\nTask: 39813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e740daf074a50d797293771ce45b1d2d35f9246e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e740daf074a50d797293771ce45b1d2d35f9246e"}]},"branch":"refs/heads/master"},"03ff4f5dc84735ff7e6ec7252caca2666d3a761c":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-05-21 07:55:12.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/729622/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/729622/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/729622/3"}}},"commit":{"parents":[{"commit":"edf60aac7bb96c7e9f91de4c19d6379792389561","subject":"Python 3: fix Ansible Vault password helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edf60aac7bb96c7e9f91de4c19d6379792389561"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 14:37:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 07:55:12.000000000","tz":0},"subject":"CI: Encrypt passwords.yml in overcloud host configure job","message":"CI: Encrypt passwords.yml in overcloud host configure job\n\nThis gives us some test coverage with Ansible Vault.\n\nChange-Id: Ic9db7f84e9f0dcbe69c31b6fff6bda3dfb401a33\nStory: 2004959\nTask: 39813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03ff4f5dc84735ff7e6ec7252caca2666d3a761c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/03ff4f5dc84735ff7e6ec7252caca2666d3a761c"}]},"branch":"refs/heads/master"},"dff50f84dd9385c38b694ceaab7540c555e7584d":{"kind":"REWORK","_number":4,"created":"2020-05-21 08:54:57.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/729622/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/729622/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/729622/4"}}},"commit":{"parents":[{"commit":"edf60aac7bb96c7e9f91de4c19d6379792389561","subject":"Python 3: fix Ansible Vault password helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edf60aac7bb96c7e9f91de4c19d6379792389561"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 14:37:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 08:54:54.000000000","tz":0},"subject":"CI: Encrypt passwords.yml in overcloud host configure job","message":"CI: Encrypt passwords.yml in overcloud host configure job\n\nThis gives us some test coverage with Ansible Vault.\n\nChange-Id: Ic9db7f84e9f0dcbe69c31b6fff6bda3dfb401a33\nStory: 2004959\nTask: 39813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dff50f84dd9385c38b694ceaab7540c555e7584d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/dff50f84dd9385c38b694ceaab7540c555e7584d"}]},"branch":"refs/heads/master"},"d69b5dff864535b3a169a3e64645f68846cd4a19":{"kind":"REWORK","_number":5,"created":"2020-05-21 09:42:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/729622/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/729622/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/729622/5"}}},"commit":{"parents":[{"commit":"edf60aac7bb96c7e9f91de4c19d6379792389561","subject":"Python 3: fix Ansible Vault password helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edf60aac7bb96c7e9f91de4c19d6379792389561"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 14:37:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 09:42:40.000000000","tz":0},"subject":"CI: Encrypt passwords.yml in overcloud host configure job","message":"CI: Encrypt passwords.yml in overcloud host configure job\n\nThis gives us some test coverage with Ansible Vault.\n\nChange-Id: Ic9db7f84e9f0dcbe69c31b6fff6bda3dfb401a33\nStory: 2004959\nTask: 39813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d69b5dff864535b3a169a3e64645f68846cd4a19"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d69b5dff864535b3a169a3e64645f68846cd4a19"}]},"branch":"refs/heads/master"},"51ca0915481d3fd40421c820da2a7afd4d834c36":{"kind":"REWORK","_number":6,"created":"2020-05-21 10:13:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/729622/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/729622/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/729622/6"}}},"commit":{"parents":[{"commit":"edf60aac7bb96c7e9f91de4c19d6379792389561","subject":"Python 3: fix Ansible Vault password helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/edf60aac7bb96c7e9f91de4c19d6379792389561"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 14:37:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 10:13:33.000000000","tz":0},"subject":"CI: Encrypt passwords.yml in overcloud host configure job","message":"CI: Encrypt passwords.yml in overcloud host configure job\n\nThis gives us some test coverage with Ansible Vault.\n\nChange-Id: Ic9db7f84e9f0dcbe69c31b6fff6bda3dfb401a33\nStory: 2004959\nTask: 39813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51ca0915481d3fd40421c820da2a7afd4d834c36"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/51ca0915481d3fd40421c820da2a7afd4d834c36"}]},"branch":"refs/heads/master"},"baf00e53e48af24b9de643aff835e335db218190":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2020-05-28 09:59:37.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/729622/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/729622/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/729622/7"}}},"commit":{"parents":[{"commit":"49c3e1dab45294a6d658746b5ca1d539d7048b55","subject":"Python 3: fix Ansible Vault password helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49c3e1dab45294a6d658746b5ca1d539d7048b55"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 14:37:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-28 09:59:24.000000000","tz":60},"subject":"CI: Encrypt passwords.yml in overcloud host configure job","message":"CI: Encrypt passwords.yml in overcloud host configure job\n\nThis gives us some test coverage with Ansible Vault.\n\nChange-Id: Ic9db7f84e9f0dcbe69c31b6fff6bda3dfb401a33\nStory: 2004959\nTask: 39813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/baf00e53e48af24b9de643aff835e335db218190"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/baf00e53e48af24b9de643aff835e335db218190"}]},"branch":"refs/heads/master"},"3b03bfe3fd52a7d2ef3231629eb96957ec551295":{"kind":"REWORK","_number":8,"created":"2020-05-28 12:10:30.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/22/729622/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/22/729622/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/22/729622/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/22/729622/8"}}},"commit":{"parents":[{"commit":"49c3e1dab45294a6d658746b5ca1d539d7048b55","subject":"Python 3: fix Ansible Vault password helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/49c3e1dab45294a6d658746b5ca1d539d7048b55"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 14:37:22.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-28 12:10:24.000000000","tz":0},"subject":"CI: Encrypt passwords.yml in overcloud host configure job","message":"CI: Encrypt passwords.yml in overcloud host configure job\n\nThis gives us some test coverage with Ansible Vault.\n\nChange-Id: Ic9db7f84e9f0dcbe69c31b6fff6bda3dfb401a33\nStory: 2004959\nTask: 39813\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b03bfe3fd52a7d2ef3231629eb96957ec551295"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3b03bfe3fd52a7d2ef3231629eb96957ec551295"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
