)]}'
{"id":"openstack%2Fkayobe~729923","triplet_id":"openstack%2Fkayobe~master~I25c7282c1e8dcdee3e7feddef9d66ca5beeb1bce","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I25c7282c1e8dcdee3e7feddef9d66ca5beeb1bce","subject":"Don\u0027t disable SELinux if it is not installed","status":"MERGED","created":"2020-05-21 11:31:45.000000000","updated":"2020-06-08 07:32:23.000000000","submitted":"2020-06-05 17:58:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"729923-1591379910532-0e537f8a","meta_rev_id":"043845bc81a4aa5474d2886d0ad8e8fd10107c77","_number":729923,"virtual_id_number":729923,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"date":"2020-05-29 13:47:27.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 17:58:30.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-05 09:22:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-04 15:17:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-06-05 09:22:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 15:17:15.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-05 09:22:43.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-05 17:58:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72a7fdafb7f71734c7ef1c9dd90087d2463a294f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-21 11:31:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3441b2c9b21d0ae161d58af1365bb3d01b42b727","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-21 13:27:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cd3f64547e6486a904256c355684f88 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/babc09b9b3634016864655fef9fa0903 : SUCCESS in 4m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/65711c3246b54c90b7440a5c60cfc309 : SUCCESS in 5m 53s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8a93f60ee2a4413380dc255273b591f6 : SUCCESS in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5dba334e247493a9f2dba7249955200 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d89151799184f579e532f02efeb2309 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/61036cdbc8b14ae69b96092662cb6338 : SUCCESS in 4m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/97774175ec7741b3a5ac588e4ff3cb7a : SUCCESS in 8m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/4367c8eec2c5428b8452d71c9e78d01b : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7dcaf371ab7a47059914b47a67608ff6 : SUCCESS in 35m 47s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d0cb93dd1297460bbb57f0a4c9077970 : SUCCESS in 1h 03m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d972c27928c4423186863e9023dc9a6f : SUCCESS in 15m 28s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/51ade5a30c88415bb7a28c7fb25ac5d6 : SUCCESS in 1h 54m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/188e852db22746efb221cf99fd3d97c4 : SUCCESS in 19m 14s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/dd7c4257044347d1b57b5f038ee68226 : SUCCESS in 25m 48s","accounts_in_message":[],"_revision_number":1},{"id":"f8af9441d58d51b3d3be4877236ca84e10018f75","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 18:33:03.000000000","message":"Patch Set 1: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit d36110ed740e711ce77362b9c57925394d233cd0","accounts_in_message":[],"_revision_number":1},{"id":"baf5c1c9eb89607b3ac17ac660514533a7e4217b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 18:34:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"c005576192e6b25098f2eee541d61c1d979da868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 20:43:35.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1c0f97918964863a363969d82d83765 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/414e1b7c9b9445e1a497464513680a54 : SUCCESS in 4m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3df1a63e9b224b78813a49c4b9d3c367 : SUCCESS in 7m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4075ef53ee694a5c8ee0ce315e8f3334 : SUCCESS in 7m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef7e6333499243fb85ec8d8e54b0107e : SUCCESS in 7m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a683020143574650ae087a7260042b75 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de213f41e5464f35857b78809117ca5d : SUCCESS in 6m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/55b938506b7a4d65bf0f0f2a06f6d8c1 : SUCCESS in 5m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/36e5201ef53343ff9dec033090485318 : SUCCESS in 11m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/36a4f882d6024725b2f9bde62c04fd36 : SUCCESS in 38m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/62ad4ad951ac4bc896a9d0358b144525 : TIMED_OUT in 2h 01m 04s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e55fcb07b0414a5b87dd27db272241e2 : SUCCESS in 20m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/340dc282a4e5415faba08e8738279d41 : SUCCESS in 1h 55m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ee86cf3cf09345589218716cec41670c : SUCCESS in 19m 36s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e8c4db9ba8ad40d889a9c6fb75e37252 : SUCCESS in 37m 34s","accounts_in_message":[],"_revision_number":2},{"id":"222a6791f8167fbf541c85f8c6ffa29d04cf32c9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-29 13:47:27.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"7caaf444f54441311a9528c95edebed7e25545da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 15:46:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e72a75765b804888896bde4a571c81dd : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffe65c79dca4408a9eb0f8eab9796294 : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e60c2f9d8b947cdac3733b1a8615013 : SUCCESS in 4m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f8b55edd5d094ed497eaab180b298084 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75a752c89f394aaba45fef4bc94718b0 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1abf8d3ac0714e3190e5b33f219a35b4 : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c801bf09eea74dfa885748b83c556c0f : SUCCESS in 3m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a606fc3bbb094c6a969eafbae5d71465 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e936a52f5a194492b2071a6d10f0f263 : SUCCESS in 6m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2eaa948f578e40e1a40f6a8ba7617d52 : SUCCESS in 26m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6cf9cecdb3ad4e06ab0ea4eb8ab24be6 : SUCCESS in 1h 07m 37s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b4ccbb11fa3c401987fb76ae42f35080 : SUCCESS in 14m 41s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/18623a0e1df04fbdaa2a48d4ad82603f : SUCCESS in 1h 58m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ca532d72a0614eac9dc0b06a722b4cc4 : SUCCESS in 16m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e4bca773cf6d406780fa72e858d85ba1 : SUCCESS in 22m 17s","accounts_in_message":[],"_revision_number":2},{"id":"b526bcc4c153173e4b1ffc9db2307da06dd152e1","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-04 15:17:15.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"40594dc77cb8c3a02ba72f3b3ef24df6f4f1b24c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-05 09:22:43.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"6a6adae770271c3f734748bc56d9fac2a2d770ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 09:22:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8c58006ef74e6d5c5d63f6c2a889a353fac8d5eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 13:00:21.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de7657ab0b484500820cd51752136c4a : SUCCESS in 6m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e41a5c2213f74f69970ddb61a993a595 : SUCCESS in 4m 32s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/f911c7998a7d4fd49ab55b2c1008d160 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c59598c28af4d49a193073cf632cb9e : SUCCESS in 5m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e481e2deef04b17a5f5df728bf610a1 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3cc9cf90295840adaf555b18bc1b9c49 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0c4c45e849fb4f8bb7176ba7e68cfbbe : SUCCESS in 7m 58s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/bb943659ef824d249d0c54f25bebed6a : SUCCESS in 34m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d8d92e33ce0740498de5d410a05e39a2 : FAILURE in 49m 51s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e6538e9fbab8443da3022aef88d3620d : SUCCESS in 17m 44s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6a0f8c4e576f4c4a9ce86b45326361b6 : SUCCESS in 2h 09m 09s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/684d85dbfeb64d66b79a0155dd0faffa : SUCCESS in 16m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9499148e66fc4fe8b5df2f55d528ab98 : SUCCESS in 30m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c903e7f0f0dca66e9384f50de27970a7a11bad08","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-05 13:03:42.000000000","message":"Patch Set 2:\n\nrecheck\n\nError: operation wait on node rabbit@centos-8-airship-kna1-0016979963 timed out. Timeout value used: 10000","accounts_in_message":[],"_revision_number":2},{"id":"73d4b5abb0046045e086a882935409c63821682d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:15:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/862388f7409d40ab8841c95745556c04 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/579a3d0d04a7431984954590f37db99a : SUCCESS in 5m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/287ee69e42fa4af49191c62267fc9853 : SUCCESS in 5m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4bddbab8e9224a73a3bb8f3ceeb91289 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87bee39f346e4295b2b8a2e5a0e74377 : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd62389af3f341c7bdb4f724aae357db : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62f75dd374454fe49bcf8ee80c81bddd : SUCCESS in 3m 16s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/37a38354a17b4b5296d4a64f7f14dc91 : SUCCESS in 5m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cca3d465532f4111bbe61dcccd830dbc : SUCCESS in 8m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f37e1b17b17a4e87bde8476604a3c76b : SUCCESS in 34m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e58be97812374f569408e78321d3539f : SUCCESS in 1h 13m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/dbf1efdc10bb4426afc7df41abbafaa4 : SUCCESS in 15m 53s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3c3fce4013d24bd7996b8308d8a7d49b : SUCCESS in 2h 04m 47s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b138be225dc64902af493ed8baf78a69 : SUCCESS in 18m 50s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/57f5d35b3563434b9085c238472cd523 : SUCCESS in 32m 14s","accounts_in_message":[],"_revision_number":2},{"id":"c002a78cedeb585212744fa57047913b7a2c4cbb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:16:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"36e0f3a5ec74f14be70b114c3cadaf971f90ddcf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 17:58:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"b8664ebe231695a724a27bc24b3b276bbe963bdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 17:58:30.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62ec85347614405ea4da3fd13cbabb28 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd00ab577bf34146a909ed08290ea11a : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22c291dc5a1c4cd6895b20ff87ce28f7 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80965adbcb284256a9f2c6a10b45ceee : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab7d40b0bd28467fb10dab3d1b49029e : SUCCESS in 3m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/46f979dc3f0440849696c0f38d95ea5b : SUCCESS in 4m 51s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a852c89b9671459784588f804969c3cf : SUCCESS in 8m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/50043342a4b943b0a4362aaa82498bab : SUCCESS in 27m 26s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8de1aaf8bfdd47ec9eae49be8435451c : SUCCESS in 1h 10m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ebfebc33fc1f43efb8350e7f7986eecb : SUCCESS in 17m 06s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6993f556b84246db80885342678edc69 : SUCCESS in 2h 02m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2763d7a6cbbd4a6eae826411bc0cc518 : SUCCESS in 18m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4e8b5db25df54168ac1f5fa0d1252c1b : SUCCESS in 28m 44s","accounts_in_message":[],"_revision_number":2},{"id":"04a551e738f7c534234d7a70c0d34e6bce972bd7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 18:00:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b28b6a2de2a4696a61d96f0713f89f0 : SUCCESS in 1m 24s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b116f846b7584e1885a51d89ff7e8ade : SUCCESS in 1m 47s","accounts_in_message":[],"_revision_number":2},{"id":"043845bc81a4aa5474d2886d0ad8e8fd10107c77","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 07:32:23.000000000","message":"Patch Set 2: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 796fed1332c44c26894a857102389f3b2f1ec8d1","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"09754df82fdc33d374877991bc73d0dfaf3c412d","revisions":{"ca19c172d7e61341a16ca9fdf79fdf3a9d150de6":{"kind":"REWORK","_number":1,"created":"2020-05-21 11:31:45.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/729923/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/729923/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/729923/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/729923/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/729923/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/729923/1"}}},"commit":{"parents":[{"commit":"26726a72ab1444cd8f7eb8eeea2a5c6c316b1871","subject":"Install libselinux-python on Ansible control host","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/26726a72ab1444cd8f7eb8eeea2a5c6c316b1871"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 11:28:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 11:31:27.000000000","tz":60},"subject":"Don\u0027t disable SELinux if it is not installed","message":"Don\u0027t disable SELinux if it is not installed\n\nKayobe has a role to disable SELinux. Some systems do not have SELinux\ninstalled (this can be reproduced by removing the selinux-policy package\nand removing /etc/selinux/config). This causes the selinux\nAnsible module to fail, since it can\u0027t write to /etc/selinux/config:\n\n    Please install SELinux-policy package, if this package is not\n    installed previously.\n\nThis change fixes the issue by only disabling SELinux if the config file\nexists.\n\nChange-Id: I25c7282c1e8dcdee3e7feddef9d66ca5beeb1bce\nStory: 2007704\nTask: 39820\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca19c172d7e61341a16ca9fdf79fdf3a9d150de6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ca19c172d7e61341a16ca9fdf79fdf3a9d150de6"}]},"branch":"refs/heads/master"},"09754df82fdc33d374877991bc73d0dfaf3c412d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-05-28 18:34:31.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/23/729923/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/23/729923/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/729923/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/729923/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/23/729923/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/23/729923/2"}}},"commit":{"parents":[{"commit":"f2c21143709d9fa752c2c694fb4dd64c0ff5010c","subject":"Merge \"CI: Encrypt passwords.yml in overcloud host configure job\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f2c21143709d9fa752c2c694fb4dd64c0ff5010c"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 11:28:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-28 18:34:20.000000000","tz":60},"subject":"Don\u0027t disable SELinux if it is not installed","message":"Don\u0027t disable SELinux if it is not installed\n\nKayobe has a role to disable SELinux. Some systems do not have SELinux\ninstalled (this can be reproduced by removing the selinux-policy package\nand removing /etc/selinux/config). This causes the selinux\nAnsible module to fail, since it can\u0027t write to /etc/selinux/config:\n\n    Please install SELinux-policy package, if this package is not\n    installed previously.\n\nThis change fixes the issue by only disabling SELinux if the config file\nexists.\n\nChange-Id: I25c7282c1e8dcdee3e7feddef9d66ca5beeb1bce\nStory: 2007704\nTask: 39820\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09754df82fdc33d374877991bc73d0dfaf3c412d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/09754df82fdc33d374877991bc73d0dfaf3c412d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
