)]}'
{"id":"openstack%2Fkayobe~730842","triplet_id":"openstack%2Fkayobe~master~I30233a86da0c4b753db1535f2c474b51c472b543","project":"openstack/kayobe","branch":"master","topic":"story/39852","hashtags":[],"change_id":"I30233a86da0c4b753db1535f2c474b51c472b543","subject":"Fix encoding of subprocess output","status":"ABANDONED","created":"2020-05-26 15:00:58.000000000","updated":"2020-05-29 13:45:37.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"2ba79e8c9718a6700b19ae028f7bc6121e167084","_number":730842,"virtual_id_number":730842,"owner":{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"date":"2020-05-29 09:33:34.000000000","_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},{"date":"2020-05-26 15:01:13.000000000","_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"date":"2020-05-28 10:03:46.000000000","_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2020-05-26 17:21:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 15:01:13.000000000","updated_by":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"reviewer":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"state":"REVIEWER"},{"updated":"2020-05-26 17:21:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-05-28 10:03:46.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"}],"messages":[{"id":"9b765e83d9e493355eedb38b6b3e336a77d9c390","author":{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},"date":"2020-05-26 15:00:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5295b5aba0150b86bba8245809b5cad2af55cd4f","author":{"_account_id":10068,"name":"Welcome, new contributor!","username":"welcome-message"},"date":"2020-05-26 15:01:13.000000000","message":"Patch Set 1:\n\nCongratulations, you\u0027ve proposed your first change in OpenDev.\n\nYour submission will now be tested automatically by Zuul, our gatekeeper,\nand reviewed by other friendly developers. They will give you feedback and\nmay require you to refine it.\n\nPeople seldom get their patch approved on the first try, so don\u0027t be\nconcerned if requested to make corrections. Feel free to modify your patch\nand resubmit a new change-set.\n\nPatches often take days (and sometimes weeks) to get reviewed, so be\npatient. Don\u0027t hesitate to ask for help, and answer questions about your\nwork promptly if you can. The more you get to know reviewers and get to be\nknown by them, the smoother the review and approval process will become. The\nfastest way to accomplish this is by reviewing other proposed changes\nyourself: anybody can do it, and it\u0027s a great way to learn the code base.\n\nThanks again for participating in OpenDev, we look forward to seeing you\naround.\n\nWorkflow Guide: https://docs.openstack.org/infra/manual/developers.html","accounts_in_message":[],"_revision_number":1},{"id":"ac243e0a3d2f3bbe6ebbe16f4a31c2f43d05be79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-26 17:21:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/072d1ba3fbd246e7abe369c72b5baf2e : FAILURE in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd30ef0483064ac8a515c281975ca4c1 : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/64206c4028fb4778b67e2d0811fc3986 : FAILURE in 6m 17s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0b1217270304fada8881e24c6facf17 : FAILURE in 9m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c2b1eb896dc4fd0bf76ca4d06a376d5 : FAILURE in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e74a056945bc48dab2bc8a0625343a45 : SUCCESS in 10m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8d9f45ebb96e4068803a535af2f77cbd : SUCCESS in 6m 32s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fe34fb92b1e740b19321089c811d268f : SUCCESS in 9m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/975b04b2644f4684b3bd77dbefe30d0c : SUCCESS in 33m 44s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c97904cb8dbf408eacb3f6ced66388e0 : SUCCESS in 1h 08m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2f270ac17307408bbc15cf4a81c87999 : SUCCESS in 16m 10s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e1cacb8b712841cf9ab69be43c88d6ac : SUCCESS in 2h 11m 08s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8db6fd794c3e474e94bc1c82e2ab91e2 : SUCCESS in 20m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f0e6df24fbfb46ce821f74bb565a2420 : SUCCESS in 44m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f653bc7ff8cfbd36298a0d9ad7beba67c1d85b5f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 10:03:46.000000000","message":"Patch Set 1:\n\nThanks for your contribution Sebastian. Great to see you getting plugged in upstream. I proposed an alternative solution quite recently, so it had not been approved or merged: https://review.opendev.org/#/c/729621","accounts_in_message":[],"_revision_number":1},{"id":"d54b1298d99b3471676426ce8069ba2524d1e7b4","author":{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},"date":"2020-05-29 09:33:34.000000000","message":"Patch Set 1:\n\nThanks, Mark.\n\nFeel free to close this to cleanup then.\n\nBest regards,\nSebastian.","accounts_in_message":[],"_revision_number":1},{"id":"8ff050bfbc93f2a717709a2353919028e8920a9a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-29 13:45:37.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5237ef02091208581b20661f3550bf90d7a35131","revisions":{"5237ef02091208581b20661f3550bf90d7a35131":{"kind":"REWORK","_number":1,"created":"2020-05-26 15:00:58.000000000","uploader":{"_account_id":32017,"name":"Sebastian Luna-Valero","email":"sebastian.luna.valero@gmail.com","username":"slunav"},"ref":"refs/changes/42/730842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/730842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/730842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/730842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/730842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/730842/1"}}},"commit":{"parents":[{"commit":"1fd8f323c1cef8d73ebb4e7098d10907ebc235a7","subject":"Merge \"Add support for encrypted block devices\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1fd8f323c1cef8d73ebb4e7098d10907ebc235a7"}]}],"author":{"name":"Sebastian Luna Valero","email":"sebastian.luna.valero@gmail.com","date":"2020-05-26 14:58:52.000000000","tz":120},"committer":{"name":"Sebastian Luna Valero","email":"sebastian.luna.valero@gmail.com","date":"2020-05-26 14:58:52.000000000","tz":120},"subject":"Fix encoding of subprocess output","message":"Fix encoding of subprocess output\n\nIn python 3, the output of subprocess.check_output is binary-encoded.\nMake sure it is decoded before returning as string.\n\nChange-Id: I30233a86da0c4b753db1535f2c474b51c472b543\nStory: 2007716\nTask: 39852\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5237ef02091208581b20661f3550bf90d7a35131"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5237ef02091208581b20661f3550bf90d7a35131"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
