)]}'
{"id":"openstack%2Fkayobe~731580","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Idb4e66aa1f2c973a956907d7986d1c05c7ce2e67","project":"openstack/kayobe","branch":"stable/train","topic":"story/2004959-stable/train","hashtags":[],"change_id":"Idb4e66aa1f2c973a956907d7986d1c05c7ce2e67","subject":"Python 3: fix Ansible Vault password helper","status":"MERGED","created":"2020-05-28 16:27:35.000000000","updated":"2020-05-28 20:25:15.000000000","submitted":"2020-05-28 20:23:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731580-1590697429418-2a4b48c7","meta_rev_id":"46379758ec99734e2ea57ca66d199046184b9125","_number":731580,"virtual_id_number":731580,"owner":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 20:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-05-28 18:07:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-05-28 16:33:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-05-28 18:07:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 16:33:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-05-28 18:07:03.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-05-28 20:23:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6c6650af8fdb89f70d708b82431d938e0921e543","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-05-28 16:27:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"b5a6e1ecf525ccba5412794f2b5892ea0b15ebbd","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-05-28 16:33:30.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6a5adf3ac4bc7b45cb79292119f1bc72ded0f8f2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 18:07:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5b99e0d37a1ee3766dc65f5c2f37a0e954489298","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 18:28:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26c1e204e81f4a838efbc8ae306869ce : SUCCESS in 5m 59s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5ce08c31227a42e2bb8809bd8f243d0f : SUCCESS in 5m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5b780007ff574ce7ad66358efb4e9d96 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6f9be1b44ac4d1b8d8405dc2dbd7f09 : SUCCESS in 5m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b30b5f831e714eb3b140820dd2541d1e : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81cfbedc003b4aa39898b9ec0246b84d : SUCCESS in 13m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a6f197cd4e804f659469ca13e00b4fbb : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7f84613f339240608ac694da9a9eb700 : SUCCESS in 8m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5d812ef98bf94845b4e345a92dd9deef : SUCCESS in 27m 44s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/137c55ed9c0c4c1ba4507a69626534cc : SUCCESS in 1h 15m 39s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d0a37e8cadd348bb944a72a3511c8949 : SUCCESS in 1h 21m 13s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/fd28220b1fe144bda729f7ce71f69608 : SUCCESS in 15m 25s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fe81257ddcd44751a3dc076c0861f372 : SUCCESS in 16m 26s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/bb8e3b5be1464bb7a6b2f82faa3472f2 : SUCCESS in 1h 52m 46s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/1f02d45dbbe2435d8eb3c97f06ac1f4a : SUCCESS in 20m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/58d5bdbe9e544d39baaaf673a1eaccbf : SUCCESS in 17m 27s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/fb50b3a4fd494a96b5207baa13be70da : SUCCESS in 39m 03s","accounts_in_message":[],"_revision_number":1},{"id":"b17b7b9b1d2b865a6ddd0a0e4d73f626580249eb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 18:28:38.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ce735c8022c6059bc467f21f536edfb3639f79bd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 20:23:49.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/9a5917b5a6a94a0fa4458f1d8d54a609 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd35594d171449ceb3fd1be4c775b056 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d41a65ebfeb74018ac11cf0b0b595acd : SUCCESS in 5m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ca7566dd8e93423b8a101eb0ca4b6f1c : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ab20b9c9d7d49569d3ecc3cecf73838 : SUCCESS in 6m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/359334d3d3a1431183796abdeb8752d0 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/654a90919dc44aebac7e482504cf9156 : SUCCESS in 7m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d2993e043db448d296535d41d00a9270 : SUCCESS in 25m 25s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/1e8d199cc0f444a18158b8a2164e3e20 : SUCCESS in 1h 12m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9821ccc80b914d3ea7d73a1abe45d1ff : SUCCESS in 1h 05m 31s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/6e5607bb17894d0bb6554a7ead06a660 : SUCCESS in 15m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e7473c5bdda841f79e72d8daeb98695c : SUCCESS in 13m 21s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b7b75313244044e7a721ea73621fd8dd : SUCCESS in 1h 53m 26s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/942ed3c76871464ab53110df9bb08bbd : SUCCESS in 21m 22s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5e20a60fd90b40cb90303fb28927326e : SUCCESS in 19m 47s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2899389b597c4a959af41085f68742eb : SUCCESS in 35m 25s","accounts_in_message":[],"_revision_number":1},{"id":"d0efa84c92e29b4774161682fd45e45ce6fcff19","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 20:23:49.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"de154d9121ad15cba63d6a7548a9cea97b9e09c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 20:25:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9667784029b94db8994cb92e716fbc55 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0c5415e979daa4f4d07328563c236774fb56e513","revisions":{"0c5415e979daa4f4d07328563c236774fb56e513":{"kind":"REWORK","_number":1,"created":"2020-05-28 16:27:35.000000000","uploader":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"ref":"refs/changes/80/731580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/731580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/731580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/731580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/731580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/731580/1"}}},"commit":{"parents":[{"commit":"104363fe7cf115adac9ac4ec6c323fb9f72c0086","subject":"Merge \"CentOS 8: Fix network configuration persistence\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/104363fe7cf115adac9ac4ec6c323fb9f72c0086"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-20 14:26:35.000000000","tz":60},"committer":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-05-28 16:27:35.000000000","tz":0},"subject":"Python 3: fix Ansible Vault password helper","message":"Python 3: fix Ansible Vault password helper\n\nIn Python 3, the return type of subprocess.check_output is bytes by\ndefault. It needs to be decoded to a string.\n\nPassing universal_newlines\u003dTrue to the subprocess functions results in\nfile objects opened in text mode, decoded using the system locale. A\nunit test has also been added.\n\nChange-Id: Idb4e66aa1f2c973a956907d7986d1c05c7ce2e67\nStory: 2004959\nTask: 39483\n(cherry picked from commit 49c3e1dab45294a6d658746b5ca1d539d7048b55)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c5415e979daa4f4d07328563c236774fb56e513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c5415e979daa4f4d07328563c236774fb56e513"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
