)]}'
{"id":"openstack%2Fkayobe~731593","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I25c7282c1e8dcdee3e7feddef9d66ca5beeb1bce","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I25c7282c1e8dcdee3e7feddef9d66ca5beeb1bce","subject":"Don\u0027t disable SELinux if it is not installed","status":"MERGED","created":"2020-05-28 18:33:03.000000000","updated":"2020-07-13 14:22:19.000000000","submitted":"2020-07-13 14:20:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731593-1594650029050-bfbdc47d","meta_rev_id":"80579575f81d8c3ad1f90fb0995077b12d7d3dda","_number":731593,"virtual_id_number":731593,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-13 14:20:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-07-13 12:23:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-08 10:10:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-07-13 12:23:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 10:10:30.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-13 12:23:19.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-07-13 14:20:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d061466f0fdf1fc5a0b6c490d059f8f407f05dde","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 18:33:03.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4e37382ed41d33afdd42344065fa6cf4d4ac9eba","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 18:33:29.000000000","message":"Abandoned\n\nnot ready yet","accounts_in_message":[],"_revision_number":1},{"id":"271a69ff3599ed48f84e159ffe1ae9dab689d39b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 07:31:57.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"d8909ad6579ed8571b31f28f1fc2e810b34af044","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 07:32:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"416f5d36074da06405b805bb63e9fba335dc35a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 09:44:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74da0cb3a6e645bdb1a67761435e42dd : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/02334fcaafa54d278769f93efdea86eb : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/04f8848655ed4aba93e74a8e608a7181 : SUCCESS in 5m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2c8790c20374228a1b147fbdf0a286d : SUCCESS in 5m 08s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8fbd255679346e5aaa3573479e32cec : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b43975887604c94a6cf7da5e7995dc3 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/904700895f3b4755b18ce22c5f46aac7 : SUCCESS in 3m 36s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b278c53bc4e64b30a2c3d66ceca83dbb : SUCCESS in 4m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3d0b307ea9ca42249b15bc336d42b27c : SUCCESS in 10m 22s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2c9d9cc6e401464084bf64dd1ad077a1 : SUCCESS in 32m 14s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/23d7a6728c6043d994e13ef4a88d03ee : SUCCESS in 1h 08m 31s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/03066acb06ec4fd2b95505dce71321b2 : SUCCESS in 1h 14m 58s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/2e38bdd97318430782c92b30b6c8b0f1 : SUCCESS in 14m 15s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c4d32e06b5294e44a894b9e61c521c07 : SUCCESS in 15m 35s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/8aefd2ce3af8452aa6e35088174c3cec : SUCCESS in 2h 06m 51s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/6fad4ed398e3462d9db5ae61127207c9 : SUCCESS in 28m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/deefa1f7a4be47f9b060018de31fe45c : SUCCESS in 16m 38s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6ceb4bc430c24f86b73dce1b740367da : SUCCESS in 41m 08s","accounts_in_message":[],"_revision_number":4},{"id":"a273893389ca8804b1cbc855c44ccf5be65c6d6f","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-08 10:10:30.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"7f1be38cc79075d2d2878cf80caaf814cb36daaa","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-07-13 12:23:19.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"c66569c214ea680a54782543e2152f064f131139","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 12:23:32.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"ac1e5da343b6e15e8cc762457fbe854c64232c19","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 14:20:28.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/ca6b9361d3bf42559d97c1d8d7494c3e : SUCCESS in 5m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c6d25819cf334dfb8487dd943142a5cd : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/808ccd3c6faf4d4198c5ad72643a851d : SUCCESS in 4m 53s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/465a9b7783024c379222e32b27e3fa4e : SUCCESS in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2a8be5a8ab4e04bdad29e1fb03bd1f : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/010e0ca2b53649e384108cd451e65613 : SUCCESS in 4m 57s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1c691fc9d6a34c00b7c438db13d54ac6 : SUCCESS in 4m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/696a1f90c1ad4267a63cc8523d22cfc7 : SUCCESS in 10m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/22c05e649b1e426eb9b218afb67fb6ff : SUCCESS in 38m 39s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/efc228a39e1f4caf934f8905377cbe37 : SUCCESS in 1h 18m 08s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6724277e33984a61acda8b044e83c1c9 : SUCCESS in 1h 03m 08s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/04001fad81544d5e9a5d45b0887dd911 : SUCCESS in 15m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ffab446261bc4a94b6949042e3f4b398 : SUCCESS in 18m 06s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9f48be3043b34884af58b4a35b76fe19 : SUCCESS in 1h 49m 15s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/fd87f67de9bc4da18eca58607b14c727 : SUCCESS in 22m 57s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/32a64077a711418ab44b46e9ccddc014 : SUCCESS in 18m 02s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/32f908844c3a44709107fcd8b3c43fac : SUCCESS in 43m 12s","accounts_in_message":[],"_revision_number":4},{"id":"7cea4ae5acb02f8c318f386d2761dbf9817361de","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 14:20:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"80579575f81d8c3ad1f90fb0995077b12d7d3dda","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-13 14:22:19.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c921017c7e44bdb21165da5fdaa95d : SUCCESS in 1m 31s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b1248230352345f69dd9ff533a84142e : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"fbf89c63a7d147d96cf77f199a0e8b7aa9aa1c5a","revisions":{"d36110ed740e711ce77362b9c57925394d233cd0":{"kind":"REWORK","_number":1,"created":"2020-05-28 18:33:03.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/731593/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/731593/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/731593/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/731593/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/731593/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/731593/1"}}},"commit":{"parents":[{"commit":"104363fe7cf115adac9ac4ec6c323fb9f72c0086","subject":"Merge \"CentOS 8: Fix network configuration persistence\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/104363fe7cf115adac9ac4ec6c323fb9f72c0086"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 11:28:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-28 18:33:03.000000000","tz":0},"subject":"Don\u0027t disable SELinux if it is not installed","message":"Don\u0027t disable SELinux if it is not installed\n\nKayobe has a role to disable SELinux. Some systems do not have SELinux\ninstalled (this can be reproduced by removing the selinux-policy package\nand removing /etc/selinux/config). This causes the selinux\nAnsible module to fail, since it can\u0027t write to /etc/selinux/config:\n\n    Please install SELinux-policy package, if this package is not\n    installed previously.\n\nThis change fixes the issue by only disabling SELinux if the config file\nexists.\n\nChange-Id: I25c7282c1e8dcdee3e7feddef9d66ca5beeb1bce\nStory: 2007704\nTask: 39820\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d36110ed740e711ce77362b9c57925394d233cd0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/d36110ed740e711ce77362b9c57925394d233cd0"}]},"branch":"refs/heads/stable/train"},"fbf89c63a7d147d96cf77f199a0e8b7aa9aa1c5a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2020-06-08 07:32:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/93/731593/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/93/731593/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/731593/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/731593/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/93/731593/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/93/731593/4"}}},"commit":{"parents":[{"commit":"8e612e8954780fdd3c0835453b5ad7b57810b73a","subject":"Merge \"CI: mark pytest results as a Zuul artifact\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e612e8954780fdd3c0835453b5ad7b57810b73a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 11:28:05.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 07:32:11.000000000","tz":0},"subject":"Don\u0027t disable SELinux if it is not installed","message":"Don\u0027t disable SELinux if it is not installed\n\nKayobe has a role to disable SELinux. Some systems do not have SELinux\ninstalled (this can be reproduced by removing the selinux-policy package\nand removing /etc/selinux/config). This causes the selinux\nAnsible module to fail, since it can\u0027t write to /etc/selinux/config:\n\n    Please install SELinux-policy package, if this package is not\n    installed previously.\n\nThis change fixes the issue by only disabling SELinux if the config file\nexists.\n\nChange-Id: I25c7282c1e8dcdee3e7feddef9d66ca5beeb1bce\nStory: 2007704\nTask: 39820\n(cherry picked from commit 09754df82fdc33d374877991bc73d0dfaf3c412d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbf89c63a7d147d96cf77f199a0e8b7aa9aa1c5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/fbf89c63a7d147d96cf77f199a0e8b7aa9aa1c5a"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
