)]}'
{"id":"openstack%2Fkayobe~731595","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Id1dfdfa6a597d89c42acfdd7350e16332a397083","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Id1dfdfa6a597d89c42acfdd7350e16332a397083","subject":"Install libselinux-python on Ansible control host","status":"MERGED","created":"2020-05-28 18:33:48.000000000","updated":"2020-06-05 14:40:31.000000000","submitted":"2020-06-05 14:37:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731595-1591367847205-da48b3b0","meta_rev_id":"caa736a6fe07293054c0110d142b37b1061adf4a","_number":731595,"virtual_id_number":731595,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 14:37:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-05 09:27:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-06-04 07:04:15.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-05 12:30:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-05 12:30:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-04 07:04:15.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-06-05 09:27:07.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-05 12:30:57.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-05 14:37:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f63ab380bafbb4e0af68839372decc03045ac9f4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-28 18:33:48.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e8c281d7236a1b2794342361d176a96fe30cd0ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 20:42:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf9e11ac964045b7892f38bd5a8eb00e : SUCCESS in 8m 08s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f95e0b29800d4a2fac9f0e88f0ed711f : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16a36be1ed5e4c04a8070588946ded00 : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b3381f1ad514b87a13188a2d6fb883d : SUCCESS in 6m 13s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ed11c29762344be0aea002bfd2493c52 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29e595ca81c4983bb60d9310e20f7f9 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf6cc4c60b8d4f4997fc6bfdd7c58ce1 : SUCCESS in 4m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e88d67904cdf40de884ddcbe374f5482 : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b023ad29666442fc99b7b0d22cf619ad : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5292471113b24a2e90aaa40846480ebe : SUCCESS in 42m 04s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/add4c7d7c8454710ad6a7acb13e72d23 : SUCCESS in 1h 16m 33s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/78c52126a7f34f709ace3f794216b706 : SUCCESS in 1h 17m 11s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/09a8f1afd8d94625861237719c260cf2 : SUCCESS in 16m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/fab9b9b52a0d4cec92e05bdf7c706c8f : SUCCESS in 16m 57s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/54651398f7784a6abec9676b3285690b : SUCCESS in 1h 57m 48s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/5f2c13c8baed4d9ba0934541da417b19 : SUCCESS in 13m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ba383e6ec7454de38cdc2861b025835b : SUCCESS in 31m 18s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/90716480a1524acc8b5b68fdf0eaf664 : SUCCESS in 35m 32s","accounts_in_message":[],"_revision_number":1},{"id":"36e3c9fb1ceac79fcfd7801164529a139eeae02b","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-04 07:04:15.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"59f1c36a5e5b3f731474055755c194bb80ef3621","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-05 09:27:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"989c1746485a88b8fe99be68961d8013fa4afc18","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-05 12:30:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nBackport candidate for other stable releases?","accounts_in_message":[],"_revision_number":1},{"id":"f4702e53482b6302a297bcc6e0fe73d6b5f0a170","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 12:31:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1fd1f65940c21f032d6b3c8ab13c1bf348c63fc0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 14:37:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7f7d561d9ab7454cacc2e14731178109 : SUCCESS in 5m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/17abea584f18437294e2ec5e4106777f : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/abd1411908f74658bab3a265113cdd35 : SUCCESS in 5m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/af99b12f93614618b7df458ae46a4697 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/751655f8d30b4ae88752c03af4b66c86 : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6f0402e81327439480860300fa7b4131 : SUCCESS in 3m 33s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/395e770da4cd4b56b5928762cd15c9c6 : SUCCESS in 5m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c976831749ff44f3b4d78bf87a884d05 : SUCCESS in 9m 45s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/816f2e2c5df64fc0bc42a3ba5c5eb9ea : SUCCESS in 39m 17s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e2d37dc83659427fa507bd296210d5b8 : SUCCESS in 1h 18m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e9c21a21a4684b87809ffe55e7a5958a : SUCCESS in 1h 14m 01s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/3156075d0c4b44c38685837c4214125b : SUCCESS in 16m 45s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/42caab8ec3f3478982dee35cb85eb392 : SUCCESS in 17m 26s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f02922c9fa77496c8aaad2afc93d7dd1 : SUCCESS in 2h 01m 33s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/dc668319d06245fbaf0cea7c72a700fc : SUCCESS in 18m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dc586f5af5db4f0dba3835a3379016c4 : SUCCESS in 16m 51s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/fc305d53b148456295f035661599bb0f : SUCCESS in 43m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e79d106afa4c785cfe90e67cbe47e57fe8a44a54","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 14:37:27.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"caa736a6fe07293054c0110d142b37b1061adf4a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 14:40:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/706376ebb3c44c44aeb85a77c7b867c5 : SUCCESS in 2m 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af84f5ffc0b24bf881185145526002b0 : SUCCESS in 1m 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0fd6fa7d9d09dbbfdb5810a64bb5d92cd5e44a22","revisions":{"0fd6fa7d9d09dbbfdb5810a64bb5d92cd5e44a22":{"kind":"REWORK","_number":1,"created":"2020-05-28 18:33:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/95/731595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/95/731595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/731595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/731595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/95/731595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/95/731595/1"}}},"commit":{"parents":[{"commit":"104363fe7cf115adac9ac4ec6c323fb9f72c0086","subject":"Merge \"CentOS 8: Fix network configuration persistence\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/104363fe7cf115adac9ac4ec6c323fb9f72c0086"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-21 11:24:00.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-28 18:33:48.000000000","tz":0},"subject":"Install libselinux-python on Ansible control host","message":"Install libselinux-python on Ansible control host\n\nCertain Kolla Ansible tasks need access to the Python SELinux module. On\nthe Ansible control host, this module is symlinked into the\nkolla-ansible virtualenv. The libselinux-python module needs to be\ninstalled for this to succeed. Typically this is the case, but in some\nenvironments it is not. This affects the kayobe control host bootstrap\ncommand:\n\nTASK [kolla-ansible : Ensure selinux Python package is linked into the\nvirtualenv]\nfatal: [localhost]: FAILED! \u003d\u003e {\"changed\": false, \"msg\": \"src file does\nnot exist, use \\\"force\u003dyes\\\" if you really want to create the link:\n/usr/lib64/python2.7/site-packages/selinux\", \"path\":\n\"/home/centos/kolla-venv/lib/python2.7/site-packages/selinux\",\n\"src\": \"/usr/lib64/python2.7/site-packages/selinux\"}\n\nThis change fixes the issue by installing libselinux-python on Python 2\nbased systems.\n\nChange-Id: Id1dfdfa6a597d89c42acfdd7350e16332a397083\nStory: 2007703\nTask: 39819\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fd6fa7d9d09dbbfdb5810a64bb5d92cd5e44a22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0fd6fa7d9d09dbbfdb5810a64bb5d92cd5e44a22"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
