)]}'
{"id":"openstack%2Fkayobe~731829","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ie42900d89afff40b8a12364b38de9854ed817275","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Ie42900d89afff40b8a12364b38de9854ed817275","subject":"CentOS 8: add docs and release note","status":"MERGED","created":"2020-05-29 18:35:55.000000000","updated":"2020-06-19 14:04:40.000000000","submitted":"2020-06-19 14:02:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"731829-1592575371755-93083828","meta_rev_id":"0b0de20190d189988401529d037381bef0bfbf32","_number":731829,"virtual_id_number":731829,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 14:02:51.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-06-17 09:32:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-18 21:22:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 13:50:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-09 16:27:28.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2020-06-17 09:32:04.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-06-19 13:50:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-19 14:02:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"446f9faacb68aa07595a8b504f216d8b1409b6b9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-29 18:35:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b7bf5d888f73b8d4eb6e42e2e751c1c3c3cb2bc0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-05-29 18:37:07.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"7cd5e92a030c39a339e0279b62c0108d0fe7415c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-29 18:46:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d7c0297c0a34047b66d62bb88267f70 : FAILURE in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f618ce4287f47378b4a1fb8a0cadd30 : FAILURE in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/794e6643ca834b14bf902d43f29601bb : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1},{"id":"0fa59e0b2c4eaf93031c9528da63afff1a5b0ba7","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-30 14:34:01.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nThanks for this, very helpful.","accounts_in_message":[],"_revision_number":1},{"id":"09ec2401e196da255deb5ae14a1fa4e05a0adbb5","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-05-30 14:40:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"af894a1fb2e362b902968f560ac085ed2dd7edd3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-01 09:42:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e1c939c10ee7242d4b4fd8bb26ecbfb9b3f8e0f6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-01 09:42:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f28ff49628996d9a469ec4a4e4b1f23a8c90294","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 13:52:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"d1e3115e9ab3443f374745bfdc6535ec01885dc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 14:04:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7fad1fd4a1c46a6b04a702c5e968e5c : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90da8843e8154e8dabee1456efe46729 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b5d7a10cc10b4d14a76ff1a677f687f0 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":2},{"id":"cd4ab630ab4e9af9a09469c811d0d22d8a3b8e5c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 15:06:20.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"2fca0b62e40a3f27d0d4a7a4131b9a771a93ef31","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 18:11:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7ac894f449745cf14d1db5fba9426addeb24c72d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 18:12:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"99f32a0fc24df97d261c552ff11b4a07bf19f6e6","author":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"date":"2020-06-09 16:27:28.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nUseful information, thanks","accounts_in_message":[],"_revision_number":2},{"id":"eff660d0e264bd3eacd26e55ec0530125bc06434","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-09 18:08:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"89fee161fa0f8dd4f3c16dc5226669e7097e3f42","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 15:22:00.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7dc2711153644af36d019360a8b31d37243aa64a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 15:38:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b9384ee3626c4d2792ec7d2b9da0b86d : SUCCESS in 9m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85f902f89e1a4e66b6fad43d5da295d1 : SUCCESS in 11m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/020748e9161a4c6abd4ff3f1597eba3e : SUCCESS in 7m 21s","accounts_in_message":[],"_revision_number":3},{"id":"51630d07c01fbb0cf31cc5899faed35607c6ed0d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 16:18:21.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"78cd4d04a1509f6ee22a051491a01a587c58ed76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 16:37:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91a922d4653d4bab80cf310c14bd0232 : SUCCESS in 7m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e82f9cb73e7c45598855f4a8e14916e8 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d5125a429cc4de181d40a5e195f621e : SUCCESS in 7m 08s","accounts_in_message":[],"_revision_number":4},{"id":"5b7cc7222c0f5aac63d72924110a5caf5563daf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 19:06:09.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2e6df79e8324e41ad00ff5575f3d3a5 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18be120c42e4403a9296018a97fcfda9 : SUCCESS in 8m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/635e04feadf24f80b9f5233cd9689380 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":4},{"id":"c470918b047feead5149d590925a7cd96d6eed85","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-17 09:32:04.000000000","message":"Patch Set 4: Code-Review+2\n\nThanks, near enough to my experience.","accounts_in_message":[],"_revision_number":4},{"id":"83cfcf222c8d4a979b6aa9f023ae16802b01f60a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 21:22:19.000000000","message":"Patch Set 4: Code-Review+2\n\nThis needs https://review.opendev.org/#/c/734055/ to be approved. Please review.","accounts_in_message":[],"_revision_number":4},{"id":"0ef36794e877009109ecd4f7889c9025c3c6d649","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 13:50:51.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5930080865772f0745e138528d4f4beb277c9bee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:51:42.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"358de74689ba07de150d347973171fe05d18546c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 14:02:51.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5746cfc42b54ee78cfb6f761d00fea1 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef49d459a07543649b32122aab0db716 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ab746f6f64ff416b82c3984cb74a354c : SUCCESS in 4m 34s","accounts_in_message":[],"_revision_number":4},{"id":"9d571df5c8fc5e17c87400afc7c86ce2b4038eab","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 14:02:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"0b0de20190d189988401529d037381bef0bfbf32","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 14:04:40.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb176b4344b040089b0264f2720e21e9 : SUCCESS in 1m 26s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6da4997ed50d424289a84150c5ebf5d5 : SUCCESS in 1m 27s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"47ca2fe72cf0f741b169e57cd05d8ae8411ecc0d","revisions":{"b125d3b366e06db1a31ca4ad6c955c0a35090783":{"kind":"REWORK","_number":1,"created":"2020-05-29 18:35:55.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/731829/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/731829/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/731829/1"}}},"commit":{"parents":[{"commit":"104363fe7cf115adac9ac4ec6c323fb9f72c0086","subject":"Merge \"CentOS 8: Fix network configuration persistence\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/104363fe7cf115adac9ac4ec6c323fb9f72c0086"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-29 18:35:29.000000000","tz":60},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nAlso adds information on using CentOS 8 in Train.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n(cherry picked from commit 922175ebd223a2a747dd70f78dfde54e81f3a4e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b125d3b366e06db1a31ca4ad6c955c0a35090783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b125d3b366e06db1a31ca4ad6c955c0a35090783"}]},"branch":"refs/heads/stable/train"},"0af0cec58a065ecbb13c75c6b3ada67eb2f9bf58":{"kind":"REWORK","_number":2,"created":"2020-06-04 13:52:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/731829/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/731829/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/731829/2"}}},"commit":{"parents":[{"commit":"104363fe7cf115adac9ac4ec6c323fb9f72c0086","subject":"Merge \"CentOS 8: Fix network configuration persistence\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/104363fe7cf115adac9ac4ec6c323fb9f72c0086"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-04 13:52:42.000000000","tz":60},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nAlso adds information on using CentOS 8 in Train.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n(cherry picked from commit 922175ebd223a2a747dd70f78dfde54e81f3a4e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0af0cec58a065ecbb13c75c6b3ada67eb2f9bf58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0af0cec58a065ecbb13c75c6b3ada67eb2f9bf58"}]},"branch":"refs/heads/stable/train"},"6880f08971b802cb4ebbcb0edb9dcc3f1cb8ca1f":{"kind":"REWORK","_number":3,"created":"2020-06-16 15:22:00.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/731829/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/731829/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/731829/3"}}},"commit":{"parents":[{"commit":"e541ea30deb6fc23a559bf35275f6f8578ba29e9","subject":"Merge \"Fix concurrency issues while adding SSH keys to known_hosts\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e541ea30deb6fc23a559bf35275f6f8578ba29e9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-16 15:21:48.000000000","tz":60},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nAlso adds information on using CentOS 8 in Train.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n(cherry picked from commit 922175ebd223a2a747dd70f78dfde54e81f3a4e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6880f08971b802cb4ebbcb0edb9dcc3f1cb8ca1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6880f08971b802cb4ebbcb0edb9dcc3f1cb8ca1f"}]},"branch":"refs/heads/stable/train"},"47ca2fe72cf0f741b169e57cd05d8ae8411ecc0d":{"kind":"REWORK","_number":4,"created":"2020-06-16 16:18:21.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/29/731829/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/29/731829/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/29/731829/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/29/731829/4"}}},"commit":{"parents":[{"commit":"e541ea30deb6fc23a559bf35275f6f8578ba29e9","subject":"Merge \"Fix concurrency issues while adding SSH keys to known_hosts\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e541ea30deb6fc23a559bf35275f6f8578ba29e9"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 20:00:21.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-16 16:18:10.000000000","tz":60},"subject":"CentOS 8: add docs and release note","message":"CentOS 8: add docs and release note\n\nAdds a support matrix page to documentation.\n\nAlso adds information on using CentOS 8 in Train.\n\nChange-Id: Ie42900d89afff40b8a12364b38de9854ed817275\nStory: 2006574\nTask: 39552\n(cherry picked from commit 922175ebd223a2a747dd70f78dfde54e81f3a4e9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47ca2fe72cf0f741b169e57cd05d8ae8411ecc0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/47ca2fe72cf0f741b169e57cd05d8ae8411ecc0d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
