)]}'
{"id":"openstack%2Fkayobe~732566","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~If0d73e13b5922ae071671d2c7983dda6a6c5ae5e","project":"openstack/kayobe","branch":"stable/train","topic":"story/2006574-stable/train","hashtags":[],"change_id":"If0d73e13b5922ae071671d2c7983dda6a6c5ae5e","subject":"Bump version of os-images role","status":"MERGED","created":"2020-06-02 12:32:10.000000000","updated":"2020-06-04 13:07:16.000000000","submitted":"2020-06-04 13:05:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"732566-1591275925463-0bd3b327","meta_rev_id":"15fad3cc5987a52bbaabdcc30e5515dcd436205c","_number":732566,"virtual_id_number":732566,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-04 13:05:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-02 20:22:42.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-04 11:00:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-06-04 06:55:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-06-04 08:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-06-04 11:00:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-06-04 08:02:08.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-02 12:32:10.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2020-06-04 06:55:28.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-06-04 08:02:08.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-04 11:00:24.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-04 13:05:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7eadea1db5899001d711265d8dbfa48ce3fa0966","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-02 12:32:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4c60f8716aa13738c049a28ae77ecc276128d858","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-02 12:58:41.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37ed0ac98149ebb5c0d2cee76c9bcc7ac4b7f0bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 15:01:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d3157dcef634c3f8d053066b216b352 : SUCCESS in 4m 38s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f2e410e6bc684b5f9da5109e44a0bc8a : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18c6675b9cd2488989b6f0a4676bcf59 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5428d97fbe8544159e2c7e78b34b5d22 : SUCCESS in 4m 30s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b5c4f87bbe644a828766b7c57bb5e9c3 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70fbd8c024ac42c781dc490239ce03d0 : SUCCESS in 6m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/23464dd6e32046c98f94fc71ced28371 : SUCCESS in 4m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ce8cac770ec84e0794db058b6432f8e7 : SUCCESS in 8m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/29b509e0b0dd4993b9d2905f2d1a65f7 : SUCCESS in 36m 31s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/6a10772e01194b748126403f67eca10e : SUCCESS in 1h 18m 58s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/db4da6fec39e4bb399f8b200cd3921aa : SUCCESS in 1h 05m 35s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/635f78d3f2e2489d82c3e2f39bc953da : SUCCESS in 16m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/7527b0c801be4dccb559e59dac03e2a2 : SUCCESS in 16m 32s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/7dfe36d7c92746d39d4b34198d176cdc : SUCCESS in 1h 57m 43s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/9503a04766dd46c39cbe516df0a11761 : FAILURE in 19m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/24f34591e56348cd96913f89de016d6b : SUCCESS in 18m 48s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4e40af68030a45feaa39990d1fc4c210 : FAILURE in 37m 59s","accounts_in_message":[],"_revision_number":2},{"id":"75be394e4cc5f07aea638d90d3bf5354a2851e32","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-02 15:15:35.000000000","message":"Patch Set 2:\n\nrecheck\n\nissues with openstack-train mirror","accounts_in_message":[],"_revision_number":2},{"id":"19a1e22b5d4f7482f9466dd2fbfa8357f108a652","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 17:10:28.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/989c01d348fd47a79933e76904b6cf5d : SUCCESS in 5m 15s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/7b568283bc514dd79d0e61e41b1e4939 : SUCCESS in 5m 15s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c797785019484beeb5c3f6a4393520b1 : SUCCESS in 5m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6985fdb6d9b4d5e9ca5ba9eab7a0cbd : SUCCESS in 5m 42s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f29ec58f81304eec8133320d00a662ad : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/378f02c98a514bfa8135a60779254975 : SUCCESS in 8m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4418bad341c14e6aa7332934bd838675 : SUCCESS in 5m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2fed5c3e10b74c07b63802bf12af5cc2 : SUCCESS in 10m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b8e975bb0aff4918a72ace2b953bd75f : SUCCESS in 29m 26s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/d4e5e69e3d2f42639eb96af4b2f26380 : SUCCESS in 1h 06m 28s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/900b2a7a20324a98a71eac420bd98113 : SUCCESS in 1h 09m 29s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/15d25ba54d9f477fb7bf3da50e5e8c2f : SUCCESS in 15m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/799781615edb4ac3ba508051502bd953 : SUCCESS in 16m 20s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0ffe4eac990d4543be39ccb76b5406e2 : SUCCESS in 1h 53m 10s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/b3af940c102640ee9f214effed1be82b : FAILURE in 20m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/90822d7c6c4c47048337fdcdb35e692c : SUCCESS in 17m 39s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f3030bd079bc4936b22838e904b1377b : FAILURE in 35m 41s","accounts_in_message":[],"_revision_number":2},{"id":"34e8a8f12a64b6d598ed24b47d6771993738dc6a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-02 20:22:42.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"d9b6c53f41797d81ecf6c872f8fd807a594e10fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-02 22:02:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/599dadcc0e614f55a59993708b863439 : SUCCESS in 4m 54s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/42f7cf36e2a847ab9330a4217fd536ea : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/376b7846338646239bc0ba5d27cabc5f : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1ea87bea4b34a97905c2742047c4ab4 : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/582b24a5f76c45f1a389a73e0fbdf76e : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c5c8739fbb4736a5a78bbcc23a729b : SUCCESS in 6m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0dc9f563e432421aa2b340cae0912971 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9285b5c4645b4ddd93815f04e3890691 : SUCCESS in 10m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/df6988e2188a4ddcbb57158aaf9c561f : SUCCESS in 34m 34s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/ffe7b8a65aff41bcbc1c15d48e90d717 : SUCCESS in 1h 15m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/52b8d711794e421fb4728bddc0647ae7 : SUCCESS in 1h 06m 58s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/013580708a0a4fa188ca0ed6fc79a6be : SUCCESS in 13m 59s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/74c58f9714b54dcfb2d64d846048159d : SUCCESS in 15m 59s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/b2b4f1e5edd246f6b179029751b3de14 : SUCCESS in 1h 36m 19s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a09ca4228e3345c39adab926854693c2 : SUCCESS in 25m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fa3afa556d194b10816306ef180b6e30 : SUCCESS in 17m 01s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e7ed6ba3aa5c4cd8834ae2b2465edc44 : SUCCESS in 53m 24s","accounts_in_message":[],"_revision_number":2},{"id":"96a83f4a3ebda71d3329ba959cf74178738b3877","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-06-04 06:55:28.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"2e4ecf311220d9c571186d14f287619fa0927bc7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-04 08:02:08.000000000","message":"Patch Set 2: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"579d74e8e16c15b3d9967820cc2292e8cc55b557","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-04 11:00:24.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"05b3b2c572fe36e450250f734b531b2038409eea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 11:00:36.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"26bae327943a51c29e182f26ee83a5ae69f3b2b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:05:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/5f39e0b9a570417b9acdcd2cd39683a6 : SUCCESS in 6m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ec4bb31e278b4ded90f766d571f19095 : SUCCESS in 8m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93a0854b52484f308839ab4a2dbc7585 : SUCCESS in 5m 51s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/50faf79a61164f3c865b2de1b263efcd : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7abc2af986744af98ca65da890978034 : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/90e652b77e00402694611f17d129ac10 : SUCCESS in 5m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b3d86078d375444bb5c7d76e23b819d1 : SUCCESS in 10m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/521836d62a77486792c0158a981647d0 : SUCCESS in 35m 39s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/db95226ab4ba4735aa766789915ba20b : SUCCESS in 1h 12m 57s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/71884333d163466e931ae724f4f1feb6 : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/9208dbb29e6848fbbd8e9a873125d1f4 : SUCCESS in 18m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5bbe312b95b741e8bd16e644abc1e5bd : SUCCESS in 19m 41s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/931640b588574059a03e06c680f744fc : SUCCESS in 1h 58m 58s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/5d1682775bba4f22baa796bea1409723 : SUCCESS in 25m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3df0821dc63a47fbad7ba48870a88162 : SUCCESS in 17m 45s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a6686eb9dd3340d899b150de0a6af935 : SUCCESS in 21m 04s","accounts_in_message":[],"_revision_number":2},{"id":"01aaffc8dab3bfe9e84762a85fa821b5b76e98c3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:05:25.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"c967eef5ab173749c916eb9826f8d9d474a0be35","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-04 13:07:16.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ae9da09043f4f69a2fc7685939a0937 : SUCCESS in 1m 32s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c731e8e932ca4d33d10f433d57c262a0f06ef3b7","revisions":{"38e5f3707d639691c3e01d3788feed7a993c5345":{"kind":"REWORK","_number":1,"created":"2020-06-02 12:32:10.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/66/732566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/732566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/732566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/732566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/732566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/732566/1"}}},"commit":{"parents":[{"commit":"0c5415e979daa4f4d07328563c236774fb56e513","subject":"Python 3: fix Ansible Vault password helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c5415e979daa4f4d07328563c236774fb56e513"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2020-05-31 10:18:15.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-02 12:31:54.000000000","tz":120},"subject":"Bump version of os-images role","message":"Bump version of os-images role\n\nVersion 1.9.0 of stackhpc.os-images removes the dependency\non shade and replaces it with openstacksdk.  This version\nis required for python3 support.\n\nChange-Id: If0d73e13b5922ae071671d2c7983dda6a6c5ae5e\nCo-Authored-By: Sebastian Luna Valero \u003csebastian.luna.valero@gmail.com\u003e\nStory: 2006574\nTask: 39485\n(cherry picked from commit d4120473c66313e3e6081f7c9d4e5ba0e508e95c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38e5f3707d639691c3e01d3788feed7a993c5345"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38e5f3707d639691c3e01d3788feed7a993c5345"}]},"branch":"refs/heads/stable/train"},"c731e8e932ca4d33d10f433d57c262a0f06ef3b7":{"kind":"NO_CHANGE","_number":2,"created":"2020-06-02 12:58:41.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/732566/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/732566/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/732566/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/732566/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/732566/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/732566/2"}}},"commit":{"parents":[{"commit":"0c5415e979daa4f4d07328563c236774fb56e513","subject":"Python 3: fix Ansible Vault password helper","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0c5415e979daa4f4d07328563c236774fb56e513"}]}],"author":{"name":"Stig Telfer","email":"stig@stackhpc.com","date":"2020-05-31 10:18:15.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-02 12:58:41.000000000","tz":0},"subject":"Bump version of os-images role","message":"Bump version of os-images role\n\nVersion 1.9.0 of stackhpc.os-images removes the dependency\non shade and replaces it with openstacksdk.  This version\nis required for python3 support.\n\nChange-Id: If0d73e13b5922ae071671d2c7983dda6a6c5ae5e\nCo-Authored-By: Sebastian Luna Valero \u003csebastian.luna.valero@gmail.com\u003e\nStory: 2006574\nTask: 39485\n(cherry picked from commit d4120473c66313e3e6081f7c9d4e5ba0e508e95c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c731e8e932ca4d33d10f433d57c262a0f06ef3b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c731e8e932ca4d33d10f433d57c262a0f06ef3b7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
