)]}'
{"id":"openstack%2Fkayobe~733833","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I68fb65cd3bc7c4390813747f9ac4ecebc3b4145f","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"I68fb65cd3bc7c4390813747f9ac4ecebc3b4145f","subject":"Remove newline from default public_net_name","status":"MERGED","created":"2020-06-05 13:43:41.000000000","updated":"2020-06-08 12:14:59.000000000","submitted":"2020-06-08 12:13:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733833-1591618404900-b9076ce0","meta_rev_id":"d5ce916481d9aa8e8c27fd78ee0d4fe95f97cc2b","_number":733833,"virtual_id_number":733833,"owner":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-08 12:13:24.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-05 13:54:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-08 10:23:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 10:23:38.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-08 12:13:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d973e0639729b290796d90496cf0e1ac7ca18c76","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-05 13:43:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"32d4b1437b6dcc0f5732266fa1b8a3c60d5b5c3e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-05 13:54:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"31ae49420d5fbcd1248958676b4f1fca58142c4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:30:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d6ddce7528ac4db2982a4eae664954e7 : SUCCESS in 5m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a466911cb783427cbb7a212bb60446f9 : SUCCESS in 6m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6d5abf0d23646bfa60feb2779b0833c : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f129c7b5b41e4ec587f1b602ee8126bc : SUCCESS in 7m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/829c73b8ef7c4550b648c5cd41676d22 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15801c5ca3e0437a9b44f8fc56085548 : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36f271b906314e5abcd32a114a7a3b94 : SUCCESS in 3m 31s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/11c5a13d1f6b4da083d3fcd21e27bfe7 : SUCCESS in 7m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ffd041b9a7204e55ab20ace47a8ea164 : SUCCESS in 10m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5d6a898322b4477ea8115cf9c1c26bcd : SUCCESS in 35m 13s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c9242401b0f54c898746c3083c217610 : SUCCESS in 1h 07m 02s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/38a0acf6ec6e4e339eed5085cbe4c5b6 : SUCCESS in 1h 38m 46s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/044d83a71a564c589f9d405847345a48 : SUCCESS in 24m 06s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ed084a6c3f844e71b78bb0eee50694eb : SUCCESS in 39m 24s","accounts_in_message":[],"_revision_number":1},{"id":"7d9841e5c13cabf10b1278f5dfd6a71cd9f49ee0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 10:23:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"15b4ec0bd59467c2d2bb9d8876ab21fdf8093c1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 10:23:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb093a7e79b075d330d5dc1ba7fcc858dee70f44","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 12:13:24.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"2c8979cbc0fb2bcc269b1e6558b4841e834a210f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 12:13:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/48e27ab70c0e4c388d2529b40fa5bf56 : SUCCESS in 5m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/29e9fe02f1fe4d94bfdec96608232173 : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64a827d6b1054bbb94ed7333f2f863ce : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9ed4d26a8a4e47aea7f8d25c9950c38d : SUCCESS in 6m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a71f4ffaa31a409bbbd93604dfde8da7 : SUCCESS in 6m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c1c796c16154b029b25110523c0e9f9 : SUCCESS in 5m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/79a4cc97b4cd4288a796661b2d17c5ec : SUCCESS in 5m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b22908b25e2b4a39b7f624ac30dd5e21 : SUCCESS in 8m 23s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a9ca9b0ff65a434b804e8552cdfb6389 : SUCCESS in 32m 39s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d00258dae1634b518962d9ea611741fb : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/3ddc7a5e105d43559ba8ef0e18fdd528 : SUCCESS in 1h 48m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/157e911395a240f0a2945ab08790baf7 : SUCCESS in 20m 10s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/ad3279ec09154a1cbafcf9b2bced15f0 : SUCCESS in 43m 04s","accounts_in_message":[],"_revision_number":1},{"id":"d5ce916481d9aa8e8c27fd78ee0d4fe95f97cc2b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 12:14:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a0a161e5c4445bcbbac93529f5ce994 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7296e87488c4cfdb9460d71db9a8fc9 : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1ecf3a684f31e0d4c783ab53c26a7e5a96ac42ef","revisions":{"1ecf3a684f31e0d4c783ab53c26a7e5a96ac42ef":{"kind":"REWORK","_number":1,"created":"2020-06-05 13:43:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/33/733833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/733833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/733833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/733833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/733833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/733833/1"}}},"commit":{"parents":[{"commit":"673cb9dbee3acf9d70f9b3d7938cbdaaea1f802d","subject":"Merge \"Fix ironic inspector rule creation idempotency\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/673cb9dbee3acf9d70f9b3d7938cbdaaea1f802d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-05-11 13:52:25.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-05 13:43:41.000000000","tz":0},"subject":"Remove newline from default public_net_name","message":"Remove newline from default public_net_name\n\nThe public network name is set via public_net_name, and its default\nvalue is the first item in the external_net_names list. The variable\nuses the YAML folded block scalar notation, but does not use the\nchomping indictator, which means the string is assigned a trailing\nnewline.\n\nThis change fixes the issue by adding the chomping indicator.\n\nChange-Id: I68fb65cd3bc7c4390813747f9ac4ecebc3b4145f\nStory: 2007654\nTask: 39731\n(cherry picked from commit 6beaf5b5e38c3a90651afb67899d8ee6661e2cf9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ecf3a684f31e0d4c783ab53c26a7e5a96ac42ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1ecf3a684f31e0d4c783ab53c26a7e5a96ac42ef"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
