)]}'
{"id":"openstack%2Fkayobe~733838","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Icb57b908be9c83700274c042185e4481d5ca6ea5","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"Icb57b908be9c83700274c042185e4481d5ca6ea5","subject":"CI: mark pytest results as a Zuul artifact","status":"MERGED","created":"2020-06-05 13:50:27.000000000","updated":"2020-06-05 20:27:29.000000000","submitted":"2020-06-05 20:26:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733838-1591388767262-f48fbef3","meta_rev_id":"90318a49c282dbc5c2cbc950e353bea1415bb28c","_number":733838,"virtual_id_number":733838,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-05 20:26:07.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-05 17:10:17.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-05 13:50:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-05 13:50:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 17:10:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-05 20:26:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"af1a0726ff08a427bc49deae1fadeb8819fe9bb6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-05 13:50:27.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"6cfccd0b4dc109164d271740b38050ce4f578075","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-05 13:50:34.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d6bc252e8ef0f84cfeb7ee79d0d89fc44fffacb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 16:14:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8df43dd73904af9ba8dcd09e4585a79 : SUCCESS in 7m 31s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f0a9ee0b44664674b98af782db3f8c44 : SUCCESS in 8m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bec312c323914ec2977cbfd60ecdef49 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/831db28df2334ad29887b64312510bb5 : SUCCESS in 7m 38s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/ece63b60f7ec486e81e6219248767db8 : SUCCESS in 6m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b280a5492da47338d27747922d65937 : SUCCESS in 10m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4ff9447e364c4fbd91543bf836d48aaa : SUCCESS in 6m 54s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8fe1d5429e6f4d029a4af305259e8208 : SUCCESS in 11m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/80f0ec66bfb7463293b0670a0690d874 : SUCCESS in 36m 17s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/a61d60edc16d433b840f478c6896c051 : SUCCESS in 1h 21m 42s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/50ed9a52315d42c39f3f17ea2625d622 : SUCCESS in 1h 15m 36s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/1794fd5a734f49b0befdc2cbe9c0dfd2 : SUCCESS in 16m 42s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/79ac3f97bb3645ae9e312767d36e9aed : SUCCESS in 20m 23s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/70471fcd95864c44938ccebf9fed904a : SUCCESS in 2h 08m 00s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/a43fd820e1a8431f876d98ca58380682 : SUCCESS in 15m 25s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/982b53d94f9b4fdba8e504688e9e92b4 : FAILURE in 12m 49s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f7952fd6641642d1961cf030b5981282 : FAILURE in 45m 41s","accounts_in_message":[],"_revision_number":1},{"id":"45a0d7059167a5105567bdb75676f7a38ebf8ff0","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-05 17:10:17.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"3deb6f5a1936455ebe4979212d2952961aab595f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:02:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e72e0770b774f1e9c83cd2d6b4ec8da : SUCCESS in 7m 36s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6ff8952fc5844908bf6a6d5f5b5e0046 : SUCCESS in 7m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3f0e4adce4e44da7909981906e7c01fb : SUCCESS in 7m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b38e51514bf469bb9aa529eb20703ba : SUCCESS in 5m 46s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1bfc2890ef9443608b8299412352e2bc : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44fbd13c50b744b1b323bc998cefde6b : SUCCESS in 8m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e9e0e4ce34684127b9f64cc357813638 : SUCCESS in 7m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4ad89e13f1b4704a29187e385d01182 : SUCCESS in 11m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/97e5e36c634b4ed0be8fd0e98ccbab0a : SUCCESS in 34m 58s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/7d3923993cb44b7faa8a1cb34b282b6d : SUCCESS in 1h 20m 04s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c815177336994fa095616a6c2490e0d2 : SUCCESS in 1h 18m 37s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/f3457ead1c0b4f0bac6578ccb15a8f6a : SUCCESS in 16m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ab8a67fd180e4be9b521179499966964 : SUCCESS in 17m 34s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6e90807ba2d04d0a8add214e70ae4d56 : SUCCESS in 1h 50m 21s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/229403722ce14a2f9a26d1cc7bce6f98 : SUCCESS in 25m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7ae327e1880a48f2bf1a8afe7e4497e0 : SUCCESS in 19m 35s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6005bd37b9fa4bbbb372d125ed5e5618 : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":1},{"id":"50565883a8be00fda7418a14515fb3cd971bbc02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 19:03:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c0ad975c4104d6d3aa8e3d7cc604be1acbec2d86","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 20:26:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"391f4e677cec49a4aa5466194137da1f5174c0f5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 20:26:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2183d224f0bf4a84a642b93653d0102e : SUCCESS in 5m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/75a54952428c4b9ebf75b81354ad1465 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2010782431a746f1bd5b5b43c882732e : SUCCESS in 4m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/fd34b30beba649da8878cedf078caa38 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c194334bfe154046bc406c02a1695e6a : SUCCESS in 8m 34s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e6f8887a005841069f603634a28001c0 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/81c5ec0b355640c1842ed232868dafc5 : SUCCESS in 9m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/c324610bc2744a54a214a6f760f48cd6 : SUCCESS in 43m 51s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/e1575024d8e747f5be1461aff339f5a2 : SUCCESS in 50m 45s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c7d384a7009b4881b44779839056313e : SUCCESS in 1h 04m 29s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/351fffaf47a54340b63a2e72ff854d72 : SUCCESS in 11m 27s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/000fd1cb5ae14843b32ade99cd6cebb7 : SUCCESS in 16m 33s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f2383187f68f45a08fb1b9dc7deba7a7 : SUCCESS in 1h 22m 00s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/490d75429d6c4567b6e2cf2410438818 : SUCCESS in 12m 53s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a8ce7f2fbbbf4b8685ed84fdcf01707a : SUCCESS in 12m 56s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/154a3e0e8ba24de8aa4ea3ff5c09d531 : SUCCESS in 31m 57s","accounts_in_message":[],"_revision_number":1},{"id":"90318a49c282dbc5c2cbc950e353bea1415bb28c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 20:27:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be8d86054ea241a9a278d2ec83d89590 : SUCCESS in 1m 05s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7bb215eac4607be530283e657ed57a86bcc94950","revisions":{"7bb215eac4607be530283e657ed57a86bcc94950":{"kind":"REWORK","_number":1,"created":"2020-06-05 13:50:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/38/733838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/38/733838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/733838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/733838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/38/733838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/38/733838/1"}}},"commit":{"parents":[{"commit":"5dbe32b2fd815c9257bc920a322e1c03c196eca5","subject":"Merge \"CI: Encrypt passwords.yml in overcloud host configure job\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5dbe32b2fd815c9257bc920a322e1c03c196eca5"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-24 19:12:31.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-05 13:50:27.000000000","tz":0},"subject":"CI: mark pytest results as a Zuul artifact","message":"CI: mark pytest results as a Zuul artifact\n\nChange-Id: Icb57b908be9c83700274c042185e4481d5ca6ea5\n(cherry picked from commit b49edbd2752bf7c4a6c81fa18a454e2e3f4bc1bf)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bb215eac4607be530283e657ed57a86bcc94950"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bb215eac4607be530283e657ed57a86bcc94950"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
