)]}'
{"id":"openstack%2Fkayobe~733840","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~Ie0b98e7364c1c00dc017129e86843962f924247b","project":"openstack/kayobe","branch":"stable/train","topic":"serialise-known-hosts-update-stable/train","hashtags":[],"change_id":"Ie0b98e7364c1c00dc017129e86843962f924247b","subject":"Fix concurrency issues while adding SSH keys to known_hosts","status":"MERGED","created":"2020-06-05 13:51:24.000000000","updated":"2020-06-09 11:43:54.000000000","submitted":"2020-06-09 11:40:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733840-1591702848012-59896c92","meta_rev_id":"6ae1024c3e65a525187075f98a2012834dd86f5b","_number":733840,"virtual_id_number":733840,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-09 11:40:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-08 18:09:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-05 17:10:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-08 18:09:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 17:10:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-09 11:40:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"49d925812cf970f75a6a1512c7a3b69cd999b21f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-05 13:51:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"9dc22c9b932127a17bf7b23e1d99af0f94e799fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:59:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/faa035e51a1f46b5a4175de46e8d5ee5 : SUCCESS in 7m 51s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/6834c40f3b5140c0a16c0dd54d39d111 : SUCCESS in 6m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c2aabc0544a3400da2fc1d7573452318 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ed140cb041f41048fe23cdb5f700c79 : SUCCESS in 6m 54s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/59a4de9a162e401a8c1d44ccd051b921 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dcaab7e82394671a090a7c3a481c5ff : SUCCESS in 12m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c722d876f78472ab98bf43c4ccf36bb : SUCCESS in 4m 41s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/60687ae542534161a896dbdb1fe37ada : SUCCESS in 9m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/35117ff4799b4dbea86ba7c4bed8c160 : SUCCESS in 12m 10s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/94d473079c8046ecbd009be19b882f8d : SUCCESS in 36m 23s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/278ea9b8dee0484cba080cc92a8d1c4c : SUCCESS in 1h 17m 35s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/928817d030b14be98f6083ad8572c3d9 : SUCCESS in 1h 07m 35s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/c7747206311f4a4db9357d7286f3b94a : SUCCESS in 18m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e26f410d02c14e1b824603b41d481444 : SUCCESS in 20m 09s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/f59d1d821f144cc8bdebf12c19498abc : SUCCESS in 1h 56m 13s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/9f76b046597f4d6bacaaf1fa7c5bb5f7 : SUCCESS in 25m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d4881930b0b34ee6b7ccdb0c994692f9 : SUCCESS in 20m 27s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/c23aecbbbda94057936048c674f6dc56 : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"5b9b2f3aa0534861bd347c0e2c98cc114166f0dc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-05 17:10:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"45c51b039c512f1592b5010b9cb8effd3a6d2827","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 18:09:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"53da5ef7a990ad202b8b062dd8a204b71523a5ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 18:09:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e89ca305f7a875fba3e8966c62b2d83a2bfdbf82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 20:15:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/068f59b1417b4cefadc29cc038679978 : SUCCESS in 4m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/404ab0adf75a46f2a77bd73364cd77af : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/206610f7040544dfb96a093974d28ae1 : SUCCESS in 4m 50s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de9793236ea644bc9e473cc05df4a868 : SUCCESS in 3m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9387fc187f5e4e3984322c02497d74e9 : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0bd29050192b44b494c776e3ed485e3a : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1cde1818bd4a42c5b825afd517fd8e04 : SUCCESS in 5m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1d65ceaff89d48a394a48fad520ae384 : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3444159605a549bd98e555e66212c43e : SUCCESS in 27m 45s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/be84e2aafa3a4bb1b2d3131b6a49513f : SUCCESS in 1h 16m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d5ebc9bde9884d4db3585b3e4dfd585a : SUCCESS in 1h 08m 31s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/6a81453c146a4c98994b5229f2fa365d : SUCCESS in 16m 22s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9c9e248c7be3431f80add43d8ed12100 : SUCCESS in 17m 52s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/5031ff7b53374c009e0fec8c256f9d61 : SUCCESS in 2h 01m 17s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/2a917a204f2540cc8eff5c3a3467e5cf : SUCCESS in 13m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bbe9ac732ae846cb9d0f4174a133dde1 : SUCCESS in 18m 27s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/3b2ef7db388f4c28bce066dd5c9bcdf1 : FAILURE in 25m 26s","accounts_in_message":[],"_revision_number":1},{"id":"83bead739eaf3960ef9b99902dbb39b71c2ff816","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-09 07:36:15.000000000","message":"Patch Set 1:\n\nrecheck\n\nGalaxy!","accounts_in_message":[],"_revision_number":1},{"id":"12d292cd554f4d44db958c02e9733297d042a1e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 09:29:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aad768b24583495da1ade8237bb9ad7d : SUCCESS in 8m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/606c43f2ac4e45a8a7aa088f67acb874 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3013edf7ed29471ca952eea27467d8d8 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01904b1632d04810818499d0f5707754 : SUCCESS in 8m 06s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/5759d0d45be04504bdbaab4503f94c43 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a52378cc09054ff79812099577c8d845 : SUCCESS in 8m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25ee84c7a8fb4943808af03ce1394853 : SUCCESS in 6m 10s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3694a64fe2cf4f52961d8fabc5e5dc9a : SUCCESS in 6m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/68987176f3134b34ab8e65ce87906daf : SUCCESS in 11m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ebfe92fd1cf944b18ee663e167c8ae7f : SUCCESS in 27m 26s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/c00108855e2d431e82314ce62706b7df : SUCCESS in 1h 25m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0767ade36f2d4f198cd81fa1fa520974 : SUCCESS in 1h 18m 48s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/72d2d87de1de4718bb02f04bb6cc8873 : SUCCESS in 21m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/c49968e8d069423c8a0c1af47d4ee2ef : SUCCESS in 21m 01s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a56bcf9bfa82479aa6088d89cf06df14 : SUCCESS in 1h 49m 22s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/66d72c5de8a344a29cf3cff8e009d79a : SUCCESS in 21m 15s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ca1197278ff44048b386b8746107e44b : SUCCESS in 22m 29s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e6a5e81d61f24679a9acf6b23df0aaff : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":1},{"id":"2fae110154f94570f8b50a3a0543969d86a93b07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 09:29:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9903dd496b1504c7b7a43af232acf2ee7d60a2f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 11:40:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a00cf6d0931d4d1c9f5476ca572f0047 : SUCCESS in 5m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53be23dddf30410b90f52d0c4dc6a7b1 : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25c3763ea0524e8d9d17eb8dcdd38758 : SUCCESS in 6m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b6abade8a40a425d9a5c48d6ce69c8c1 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7149d0e579444f25885daa1c593eaa4b : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9724268f5774b238555950c342371fa : SUCCESS in 3m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c850440bb5a34618ad8bb64374ebff02 : SUCCESS in 5m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/023fd30683624f90a53270c65ddedf56 : SUCCESS in 10m 13s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4989f5391944e8c9a1d66481f6e206d : SUCCESS in 28m 38s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/6dc1dedf42e840c483c0d308550fc275 : SUCCESS in 1h 18m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6f819a312c4d46a1ad5feb4cb8136ca4 : SUCCESS in 1h 08m 34s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/feb45c15988743639be6b3aad15961ee : SUCCESS in 16m 35s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/af071c2ed1ba415baa5b93570a6f9822 : SUCCESS in 18m 17s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a9ab577d3bb948d1b0ed2b61d1bcc0bf : SUCCESS in 2h 05m 35s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/434f3d2b04e9440f80b40596de3fff2a : SUCCESS in 23m 05s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/331e9ef5f8754ab1a6c30fe929b71f34 : SUCCESS in 19m 44s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/564dcbfde9354cdbb5eef4282b04e47c : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":1},{"id":"48e48b340a9320a4748971d7c407bf44f40ccc13","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 11:40:48.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"b01fee0b3203affa1cf9d89d2eb412aa37a9cb63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-09 11:43:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79fa81ce5add4f5db984f1d98f48f39d : SUCCESS in 2m 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f7a6d36ba78f45f7ad4d0b4ef60670df : SUCCESS in 1m 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"981658d5f2aaecb0eb7a7f7d35c7495d15da1847","revisions":{"981658d5f2aaecb0eb7a7f7d35c7495d15da1847":{"kind":"REWORK","_number":1,"created":"2020-06-05 13:51:24.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/40/733840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/733840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/733840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/733840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/733840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/733840/1"}}},"commit":{"parents":[{"commit":"5dbe32b2fd815c9257bc920a322e1c03c196eca5","subject":"Merge \"CI: Encrypt passwords.yml in overcloud host configure job\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5dbe32b2fd815c9257bc920a322e1c03c196eca5"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-11 13:44:18.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-05 13:51:24.000000000","tz":0},"subject":"Fix concurrency issues while adding SSH keys to known_hosts","message":"Fix concurrency issues while adding SSH keys to known_hosts\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: Ie0b98e7364c1c00dc017129e86843962f924247b\nStory: 2007628\nTask: 39672\n(cherry picked from commit c5b2d435eeb13f497fd71bd0c8b85b6ef4a9a731)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/981658d5f2aaecb0eb7a7f7d35c7495d15da1847"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/981658d5f2aaecb0eb7a7f7d35c7495d15da1847"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
