)]}'
{"id":"openstack%2Fkayobe~733842","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Ie0b98e7364c1c00dc017129e86843962f924247b","project":"openstack/kayobe","branch":"stable/rocky","topic":"serialise-known-hosts-update-stable/rocky","hashtags":[],"change_id":"Ie0b98e7364c1c00dc017129e86843962f924247b","subject":"Fix concurrency issues while adding SSH keys to known_hosts","status":"MERGED","created":"2020-06-05 13:51:50.000000000","updated":"2020-06-19 12:28:23.000000000","submitted":"2020-06-19 12:26:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733842-1592569601048-e6e5d7b1","meta_rev_id":"b12b267cac8e2c19a602bab1db5dc3daf0ba9eb6","_number":733842,"virtual_id_number":733842,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 12:26:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-19 10:37:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-19 10:05:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-19 10:37:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 10:05:44.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-19 12:26:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7edb946b50a7731c02671046f15ff65a3a4cc311","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-05 13:51:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e470434c5e9b925a54ca5897d333f512dfcfb728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 15:07:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/15d785dc116e4ffd9dd617c6dc2880cd : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fcc7a693ac944dcfa67f490530ff390d : SUCCESS in 6m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99784a0685504e42abb178fc7e0054b7 : SUCCESS in 6m 48s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c447603589c047d298b127c3a60d8d09 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70148b1963c743d0b65c7b4b395f20c7 : SUCCESS in 6m 45s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/493b048f0dcc42df8c1564480d56deb1 : SUCCESS in 6m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9478e33ead464cb986aeaec373c278cb : RETRY_LIMIT in 4m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/25c542ba96564ec9beff6e95e8208438 : SUCCESS in 8m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/590968d3d76445d899686c952328bdff : SUCCESS in 9m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e602d6678aff49d6a4fcc6de9da0088d : SUCCESS in 26m 29s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/dbcfce2f39ae4f40ac81b38f75d1778a : SUCCESS in 1h 04m 40s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/612345aac7264319b18b528f97e51ad2 : FAILURE in 11m 14s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/806b235ebbd74d75a2220379efc4aaff : SUCCESS in 23m 04s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/91f8f609c0984020a526cf4d91ba5110 : FAILURE in 11m 16s","accounts_in_message":[],"_revision_number":1},{"id":"cc54a970f267fbce5be1379d0ef584b70d83dc7d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 14:14:36.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"8d369e83c6d3d08e2a09fc8970ca4664c5d585bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:31:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/f7326b93c8934bb4aa9e7c2750aa2de6 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6c302421ad45f78e2a512e5254a8f5 : SUCCESS in 5m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49721855953f44e29d01bda210d1de57 : SUCCESS in 5m 49s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a089d325d71643f9b15a515c2a56e840 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db74d18122eb44199b1bac655856145a : SUCCESS in 4m 04s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e0f8f3f3b4ee47fcba1942c67454a32e : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0857b15d4f864b6ea861d46c7e8cccb8 : RETRY_LIMIT in 8m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c08b4f1b960d4c8e82fa1aca84850dc2 : SUCCESS in 7m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3292576ae2c74b1ebddd8ad480891d53 : FAILURE in 7m 00s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/95f4a1668f1542e2bd7291af36bd9f6b : SUCCESS in 25m 07s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d9b6a8f47231442bb4e8a92c86a8e60d : SUCCESS in 1h 01m 13s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5039709a820f44d5996d869b29dc4670 : FAILURE in 52m 53s (non-voting)\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/15aaa957a045407895bfa5c2edf93bce : SUCCESS in 22m 01s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/13baf13e9d6042a09bb685e323354842 : FAILURE in 13m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2f29cb57ee8724481f8aacc85f9729cbd0500985","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 07:35:01.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"f659f3ea96f0810d17d59187f678ee0240cf329b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 07:35:41.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"ec4f24e314070185b42197a4cc06491955591902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 08:43:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c97b010687284f679b81920e956b2271 : SUCCESS in 7m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/33505f22984942498132f614081fdc55 : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cff06813a9ea4dc8860c6c67c1dc2ec0 : SUCCESS in 6m 43s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/4707552c0ee64670a8568c032b0815e5 : SUCCESS in 6m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ace37ab6e5e46f08ebf3515b0a9dc8f : SUCCESS in 10m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f439693f4fe541f480bd113ca892d08e : SUCCESS in 7m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/c09c0dadfe714ceaabec4e5b13f6fb74 : SUCCESS in 6m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/347bcde7a6fd4ec19916e9b6da3d1dfb : FAILURE in 6m 18s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/170f7c1365354d9aa74168e0c222c0dd : SUCCESS in 30m 32s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9cdfa7becbfd4d98b46dba8724da6d9d : SUCCESS in 1h 05m 32s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/a364e714519940199a1585d342a97e54 : FAILURE in 36m 06s (non-voting)\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/dbf591f040944806a7d445a44d5470f0 : SUCCESS in 24m 43s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/cb080547242c4063b039e3ea1ebfe7a4 : FAILURE in 16m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"121d9457eb8d86cef10a721906a8c839b44326f7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 10:05:44.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dfc45db5bab6a236005fd8271caf253f1f375ac2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-19 10:37:58.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8485ca27f50587e9028f4f86a8a0655cd78f3e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 10:38:11.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"3c3af794e9b677f9da4d407744f3afc2828d7bba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:26:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/40e0f9cc1ce647268f044d5b941ce01f : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/351ba59281cc4872a889b659e5cd399c : SUCCESS in 5m 03s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/6c0cb66787724f47a4223445093e3ba8 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5c82ac207904fa2a425e69b12b6c871 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/704dfaa71ff145409a72edfd3522c820 : SUCCESS in 3m 42s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/edd61281d3054ca2892b778922f9cb55 : SUCCESS in 4m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/409f3d09784442a1a13fcf9e44e84339 : SUCCESS in 31m 29s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/25e2ca30612445c79c13e9e254fa3f6d : SUCCESS in 1h 04m 29s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/03d85ee4c68c4642b6b3b5071cb4ba46 : SUCCESS in 24m 49s","accounts_in_message":[],"_revision_number":3},{"id":"fb885cb66df9cc0f3ae18fec68b88731c4abd94e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:26:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"c97bf9e0d0989f4310385c027e01a0760b91f8f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:28:23.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e6f8361e0514d47bff4b249d32faf35 : SUCCESS in 1m 16s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94920394d739411fba5e0fa0e7e485fe : SUCCESS in 1m 23s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"64c36f7015636a5e5423b70fb3fa576044235652","revisions":{"71c9c45c2cca26b33759a8b12b227155fed38960":{"kind":"REWORK","_number":1,"created":"2020-06-05 13:51:50.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/42/733842/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/733842/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/733842/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/733842/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/733842/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/733842/1"}}},"commit":{"parents":[{"commit":"5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59","subject":"Use ensure-docker role instead of install-docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-11 13:44:18.000000000","tz":120},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-05 13:51:50.000000000","tz":0},"subject":"Fix concurrency issues while adding SSH keys to known_hosts","message":"Fix concurrency issues while adding SSH keys to known_hosts\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: Ie0b98e7364c1c00dc017129e86843962f924247b\nStory: 2007628\nTask: 39672\n(cherry picked from commit c5b2d435eeb13f497fd71bd0c8b85b6ef4a9a731)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71c9c45c2cca26b33759a8b12b227155fed38960"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/71c9c45c2cca26b33759a8b12b227155fed38960"}]},"branch":"refs/heads/stable/rocky"},"c5d7c5572225f9e03f92b859c625f4076262d3b7":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-18 14:14:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/42/733842/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/733842/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/733842/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/733842/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/733842/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/733842/2"}}},"commit":{"parents":[{"commit":"6dd11cea73e0e2df4d65ecc3744e4abc68ff195f","subject":"CI: don\u0027t run virtualenv on an existing virtualenv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6dd11cea73e0e2df4d65ecc3744e4abc68ff195f"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-11 13:44:18.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-18 14:14:36.000000000","tz":0},"subject":"Fix concurrency issues while adding SSH keys to known_hosts","message":"Fix concurrency issues while adding SSH keys to known_hosts\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: Ie0b98e7364c1c00dc017129e86843962f924247b\nStory: 2007628\nTask: 39672\n(cherry picked from commit c5b2d435eeb13f497fd71bd0c8b85b6ef4a9a731)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5d7c5572225f9e03f92b859c625f4076262d3b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c5d7c5572225f9e03f92b859c625f4076262d3b7"}]},"branch":"refs/heads/stable/rocky"},"64c36f7015636a5e5423b70fb3fa576044235652":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-19 07:35:41.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/42/733842/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/42/733842/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/733842/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/733842/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/42/733842/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/42/733842/3"}}},"commit":{"parents":[{"commit":"7de6f8b12552fad83a7edb99254f3fa403a30c41","subject":"Remove newline from default public_net_name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7de6f8b12552fad83a7edb99254f3fa403a30c41"}]}],"author":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-05-11 13:44:18.000000000","tz":120},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-19 07:35:41.000000000","tz":0},"subject":"Fix concurrency issues while adding SSH keys to known_hosts","message":"Fix concurrency issues while adding SSH keys to known_hosts\n\nCo-Authored-By: Mark Goddard \u003cmark@stackhpc.com\u003e\n\nChange-Id: Ie0b98e7364c1c00dc017129e86843962f924247b\nStory: 2007628\nTask: 39672\n(cherry picked from commit c5b2d435eeb13f497fd71bd0c8b85b6ef4a9a731)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64c36f7015636a5e5423b70fb3fa576044235652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/64c36f7015636a5e5423b70fb3fa576044235652"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
