)]}'
{"id":"openstack%2Fkayobe~733846","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Id4d951e447720e1d769491c0d34ad83099c030eb","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"Id4d951e447720e1d769491c0d34ad83099c030eb","subject":"Use Ansible connection reset support in docker role","status":"MERGED","created":"2020-06-05 13:53:35.000000000","updated":"2020-08-06 12:16:31.000000000","submitted":"2020-08-06 12:14:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"733846-1596716087761-2e16cef6","meta_rev_id":"54477524f8c9c4494f6a82b8fa26508af2408334","_number":733846,"virtual_id_number":733846,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-06 12:14:47.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":2,"date":"2020-08-06 09:59:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-05 17:08:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"all":[{"value":1,"date":"2020-08-06 09:59:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-05 17:08:06.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-08-06 09:59:40.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-08-06 12:14:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0dda0a3e8012bcbaf175a2f2e8cee21f1cbff140","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-05 13:53:35.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c6d2f574ba1aee2dc5db203e766541a8d688a71e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-05 16:10:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/634187010a6c487e8a60b5c03d80392a : SUCCESS in 8m 03s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/1229736593184fa0a538fa8a031e3196 : SUCCESS in 6m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cb2357233e624b50b66cc2931e7e6e4b : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/57a37106bad24fef84835ea171077543 : SUCCESS in 6m 10s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/088701b75d5a4eeb97c1a475e254ef43 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3fe9d3b26844e8187bd6339b1f19d5a : SUCCESS in 6m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/191d19f9fa9942e486b9e023a4ee60d0 : SUCCESS in 6m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9572f6543bbe490fb4cf1516c723117b : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/28a62baadfa744cca5be238f10b6d327 : SUCCESS in 37m 21s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/854caac9d97e4bcb979f92d29794ab4e : SUCCESS in 1h 01m 05s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/5c931bb36f2e4a759a659d95e7c97108 : SUCCESS in 2h 00m 21s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/43323efaa7284d72a26ddc9776a60ab3 : SUCCESS in 26m 29s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/90830f49713942e1825365fc5af62332 : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":1},{"id":"6c019e0dc87f136390f9e73d01cb5f6ca5a64f79","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-05 17:08:06.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa3ddaa47243cf4dfd52ac9f7f771808fe18deca","author":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"date":"2020-08-06 09:59:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"56dc87971fd061c5c111c8523758894da0fc747e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 09:59:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5519a0a06550b534242367819eb3253422072d8a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 12:14:47.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cbdda695a75f16acd407fbbf8ba46cb0528a5b0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 12:14:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/13a6cfc6bdf04dd18b2b94ceea5c03a6 : SUCCESS in 5m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7bb44b2fa85941bca8152bbfc9206cce : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/274104fb674b4b0e8729a14cc28804bf : SUCCESS in 4m 18s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/3fb444d3955a47c09d853a1550030a8d : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a29dd9c078b4271be2448cbc259a206 : SUCCESS in 4m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/69567f506cb04a9da2f08ab6a87ed940 : SUCCESS in 5m 02s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f8264665879640b098b2eafb1499154e : SUCCESS in 8m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f3a2ebe8b2cc482a8e28c166efdd9748 : SUCCESS in 16m 15s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d65cf173d3024b3eb17eecc196fbd8cb : SUCCESS in 1h 08m 41s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/8153655a4f6d4f19b611a0ffc70eedd3 : SUCCESS in 1h 59m 17s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/567401e476d44d33b6833d1e248eb2c1 : SUCCESS in 25m 59s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/3ffc636f5ebd43c19de07fceaec3502e : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":1},{"id":"54477524f8c9c4494f6a82b8fa26508af2408334","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-06 12:16:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a65ce64d722e45fc9c52a7e0d8985692 : SUCCESS in 1m 14s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"34e1985888480ed0b8ce83cc1f4421dbe2da410d","revisions":{"34e1985888480ed0b8ce83cc1f4421dbe2da410d":{"kind":"REWORK","_number":1,"created":"2020-06-05 13:53:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/46/733846/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/46/733846/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/733846/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/733846/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/46/733846/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/46/733846/1"}}},"commit":{"parents":[{"commit":"673cb9dbee3acf9d70f9b3d7938cbdaaea1f802d","subject":"Merge \"Fix ironic inspector rule creation idempotency\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/673cb9dbee3acf9d70f9b3d7938cbdaaea1f802d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-04-03 16:25:24.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-05 13:53:35.000000000","tz":0},"subject":"Use Ansible connection reset support in docker role","message":"Use Ansible connection reset support in docker role\n\nWhen configuring Docker we need to kill persistent SSH connections to\nrefresh the membership of the docker group for the stack user. Currently\nwe are using a fairly heavy handed method of removing all ControlPersist\nsockets because the Ansible reset_connection meta module previously did\nnot work [1]. This issue is fixed since Ansible 2.5.6.\n\nThis change switches to the reset_connection meta module, which now\nworks as expected.\n\n[1] https://github.com/ansible/ansible/issues/27520\n\nChange-Id: Id4d951e447720e1d769491c0d34ad83099c030eb\n(cherry picked from commit 6ca967e27fd008db57c92e4885ff2daad52a7ff5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34e1985888480ed0b8ce83cc1f4421dbe2da410d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34e1985888480ed0b8ce83cc1f4421dbe2da410d"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
