)]}'
{"id":"openstack%2Fkayobe~734040","triplet_id":"openstack%2Fkayobe~master~Ib6ab41a3ba320a1fe15d0d23561fad2fab7861e6","project":"openstack/kayobe","branch":"master","topic":"story/2007259","hashtags":[],"change_id":"Ib6ab41a3ba320a1fe15d0d23561fad2fab7861e6","subject":"Fix seed VM interface ordering","status":"MERGED","created":"2020-06-08 09:06:48.000000000","updated":"2020-06-15 11:49:50.000000000","submitted":"2020-06-15 11:46:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734040-1592221614188-a663f06a","meta_rev_id":"de333973c62225cb032fa72a3dbd0a55d5da6fb2","_number":734040,"virtual_id_number":734040,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-15 11:46:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-06-15 09:08:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-15 09:53:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-15 09:53:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-06-15 09:08:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-06-15 11:49:50.000000000","post_submit":true,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-06-15 11:49:50.000000000","post_submit":true,"permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"value":0,"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 09:07:15.000000000","updated_by":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"reviewer":{"_account_id":10910,"name":"Stig Telfer","email":"stig.openstack@telfer.org","username":"stigtelfer"},"state":"REVIEWER"},{"updated":"2020-06-15 09:08:55.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-06-15 09:53:38.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-15 11:46:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7e715824e2a2f5888f842a30e9689c7216be4663","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 09:06:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7725abd3e851e0852cd90704828b272fc92dc021","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 09:42:57.000000000","message":"Topic set to story/2007259","accounts_in_message":[],"_revision_number":1},{"id":"a2f182ab1ebd4005ffcbdb10f5339e362f7809e5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 09:42:59.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"ed6c2d92db874cdd9fae2385af9278df3d70de65","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 09:46:39.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"682766fae78626e26705dae2399009bb74f66379","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 09:48:05.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"e1050baf72edb8c5796c723ac55e4c5240c9ec3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 11:37:42.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e992d0d8eb64d70b0209e26791fa060 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a454e368a6241b38897496abc014bac : SUCCESS in 5m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b757d6e8c957487f884571242201f9b3 : SUCCESS in 5m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c2e5c931ec864ba291380f734c220d67 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/098a7b30ad574904a61397118ed46ca4 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/141d1702d0e244ff8110688737291879 : SUCCESS in 7m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcc4d4e1c08f47e6abce37c50d9157fc : SUCCESS in 3m 25s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0a795dc49df34ae4863265780bdac861 : SUCCESS in 5m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8a1ffffbabd94e239ff1bd037fef1517 : SUCCESS in 9m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e7959039534b43508be622c4b01d7858 : SUCCESS in 26m 09s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/db3ab9390acc438c868d339e05470847 : SUCCESS in 1h 03m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/31bc8fbd3e0f4aacbcb3eb83fa3755e9 : SUCCESS in 15m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fea50f29d81a479cbee883c336a64977 : SUCCESS in 1h 48m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/dd53c6663d8146bf8de06cb50a0b307f : SUCCESS in 15m 52s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3ae65f87bb474f63826b7b9d5520150e : SUCCESS in 25m 26s","accounts_in_message":[],"_revision_number":4},{"id":"2f554c0b17fd4dd6eb00e346d0e377af906a791d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-10 15:30:13.000000000","message":"Patch Set 4: Review-Priority+1","accounts_in_message":[],"_revision_number":4},{"id":"76828e3c733462979754ddea8e930f24efd00226","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-11 17:44:22.000000000","message":"Patch Set 4: Backport-Candidate-1\n\n(1 comment)\n\nCode looks good, comments about docs.","accounts_in_message":[],"_revision_number":4},{"id":"fc3f842095124c80f7099cfddd97d9967a5f811a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-12 16:32:58.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bd7cf8b5332dd972bd8567092fd8dce49482bf22","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-12 16:33:21.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"97b9d38f4ecf0838cab29f2458b6779c30f54743","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 18:25:30.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5edabeb8ce1f418e9e9126de7e4ace7b : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/016d4537d27c400cae677c677f8cd6b4 : SUCCESS in 4m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8e23d120248a43beba6b894ddca706f5 : SUCCESS in 4m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/539f2a69be404b2baf6b31bf29c676ec : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ae923135d6e41dfb1abbdf5e8e8df4f : SUCCESS in 9m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/41555279bf254bb4b9add53676450a90 : SUCCESS in 3m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/36e3ff333c14415987672ddd42b90803 : SUCCESS in 5m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/21ddaeb5b32f4e57823ebd55f207de40 : SUCCESS in 8m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1154902c019e4244b6ce0e3be36bb2bf : SUCCESS in 32m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/07a2d5614a48452480b27747ff8eac55 : SUCCESS in 1h 13m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0ddd017d266a4ebd9079c6521498c646 : SUCCESS in 17m 04s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/10eb5c3def9344e18e6eab19092ea23f : SUCCESS in 1h 47m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/08036dd969584ef4923260a87e372be1 : SUCCESS in 25m 24s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b0b063d03874425d865fb8f11a365a0a : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":5},{"id":"61a30d28560a2d8471dd74a40a6f072ade9f7db9","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-15 09:08:55.000000000","message":"Patch Set 5: Backport-Candidate+2 Code-Review+2\n\nThanks for fixing this. Perhaps we could add some more notes when we do an upgrade and need to figure out how to configure this mapping.","accounts_in_message":[],"_revision_number":5},{"id":"bbb7e196f768adc4079817b8a9a05770254c9669","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-15 09:53:38.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\nThank you Mark for the revised patch.","accounts_in_message":[],"_revision_number":5},{"id":"0b715336d817cf5c35fe61c6677fa012a706db79","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 09:59:22.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"d1bd78a042ecd25d242f5a7498f750d73c3da1ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 11:46:53.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1070bf97684408a81f9eccc737dc21a : SUCCESS in 4m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c537461e2bbe4fec9b55a9d5cfb0f894 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b24e188883f848fe87d48e57a27d9278 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b01bd6b740dc4213bc771433c1ddd504 : SUCCESS in 8m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d95ee145472c476fbc07e85c004aaf23 : SUCCESS in 4m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b240955c5abe4f3e884f7e72498d3183 : SUCCESS in 5m 44s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/94ef6cbc39304b61a535fd22fcdbf074 : SUCCESS in 9m 38s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1f2557f528eb4f4e928066698434360f : SUCCESS in 35m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/659219856be148b0a3e79f27e421f300 : SUCCESS in 1h 05m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/85473fc8f0904e30ba1623dd96d552b2 : SUCCESS in 17m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f3aef7e553f04efd910981d351a32243 : SUCCESS in 1h 43m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5ac5175da356472b893bc9b5d8e9c853 : SUCCESS in 19m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7cc90760c79a43eaa18442a20c85714c : SUCCESS in 28m 31s","accounts_in_message":[],"_revision_number":5},{"id":"a31819f0bee86b52a665681c42ccbeb5a14ad0e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 11:46:54.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"cbb9c68c944a77c964f1378915c9a305e95559fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 11:48:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08f9c77f1f5e44419e8a14f0852dbcd4 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c8f17cb1351457cbfe25722007faaba : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":5},{"id":"b0cc22eeb0d7a67a268586e855de080c4abfc98a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-15 11:49:32.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit d918046a8ee1f291213e2d1b35ae28455809f513","accounts_in_message":[],"_revision_number":5},{"id":"6c8eac210b0bfa525fe8cbe129cbaa9329bf5cd0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-15 11:49:50.000000000","message":"Patch Set 5: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 847bd9e99272e9668a06ee28a78de146ba7dc12e","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7bcc5579e2de1735deaf970c5093a4ce629a37f0","revisions":{"ec3b325622687983d2d26caab9d8a22fa8c7f48b":{"kind":"REWORK","_number":1,"created":"2020-06-08 09:06:48.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/40/734040/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/734040/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/734040/1"}}},"commit":{"parents":[{"commit":"c2f049c6f01ee5d6dd383a5edef7fc051f79447f","subject":"Merge \"[Community goal] Update the contributor guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2f049c6f01ee5d6dd383a5edef7fc051f79447f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 08:53:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 09:06:37.000000000","tz":60},"subject":"Fix seed VM interface ordering","message":"Fix seed VM interface ordering\n\nCurrently the ordering of network interfaces in the seed VM is\nnon-deterministic. This happens because we apply the \u0027unique\u0027 filter to\nthe network_interfaces list, which does not guarantee a deterministic\nordering. This list is then transformed and passed to the\nstackhpc.libvirt-vm role.\n\nThere are two consequences of this:\n\n* it is not possible to determine which interface names should be used\n  prior to creating a seed VM\n* if a seed VM is recreated, the interface ordering may change\n\nThis change fixes the issue by sorting the network_interfaces list\nalphabetically before it is transformed and passed to the\nstackhpc.libvirt-vm.\n\nA new \u0027seed_vm_interfaces\u0027 variable is also added, which allows for\ncustomisation of the VM\u0027s interfaces - potentially allowing for more\ncomplex setups such as trunked VLANs.\n\nChange-Id: Ib6ab41a3ba320a1fe15d0d23561fad2fab7861e6\nStory: 2007259\nTask: 38621\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec3b325622687983d2d26caab9d8a22fa8c7f48b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec3b325622687983d2d26caab9d8a22fa8c7f48b"}]},"branch":"refs/heads/master"},"ff23cc7e8c732f498d952f5b23793c59e6377abd":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-08 09:42:59.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/40/734040/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/734040/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/734040/2"}}},"commit":{"parents":[{"commit":"c2f049c6f01ee5d6dd383a5edef7fc051f79447f","subject":"Merge \"[Community goal] Update the contributor guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2f049c6f01ee5d6dd383a5edef7fc051f79447f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 08:53:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 09:42:42.000000000","tz":0},"subject":"Fix seed VM interface ordering","message":"Fix seed VM interface ordering\n\nCurrently the ordering of network interfaces in the seed VM is\nnon-deterministic. This happens because we apply the \u0027unique\u0027 filter to\nthe network_interfaces list, which does not guarantee a deterministic\nordering. This list is then transformed and passed to the\nstackhpc.libvirt-vm role.\n\nThere are two consequences of this:\n\n* it is not possible to determine which interface names should be used\n  prior to creating a seed VM\n* if a seed VM is recreated, the interface ordering may change\n\nThis change fixes the issue by sorting the network_interfaces list\nalphabetically before it is transformed and passed to the\nstackhpc.libvirt-vm.\n\nA new \u0027seed_vm_interfaces\u0027 variable is also added, which allows for\ncustomisation of the VM\u0027s interfaces - potentially allowing for more\ncomplex setups such as trunked VLANs.\n\nStory: 2007259\nTask: 38621\n\nThere is a second issue, which is that if the seed VM has a\nnetwork interface not configured with a gateway, cloud-init will fail to\nconfigure the network interfaces on the host. This has been observed on\nCentOS 8, but is probably more tied to the version of cloud-init, and\nmay affect CentOS 7. The following error is seen in the cloud-init logs:\n\n    KeyError: \u0027gateway\u0027 \n\nThis change has been addressed in the jriguera.configdrive role, and\nthis change updates the version used in requirements.yml.\n\nStory: 2007769\nTask: 39993\n\nChange-Id: Ib6ab41a3ba320a1fe15d0d23561fad2fab7861e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff23cc7e8c732f498d952f5b23793c59e6377abd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ff23cc7e8c732f498d952f5b23793c59e6377abd"}]},"branch":"refs/heads/master"},"c3b3e1826b28d6eac611e28378f5cdf815703f46":{"kind":"REWORK","_number":3,"created":"2020-06-08 09:46:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/40/734040/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/734040/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/734040/3"}}},"commit":{"parents":[{"commit":"c2f049c6f01ee5d6dd383a5edef7fc051f79447f","subject":"Merge \"[Community goal] Update the contributor guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2f049c6f01ee5d6dd383a5edef7fc051f79447f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 08:53:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 09:46:29.000000000","tz":60},"subject":"Fix seed VM interface ordering","message":"Fix seed VM interface ordering\n\nCurrently the ordering of network interfaces in the seed VM is\nnon-deterministic. This happens because we apply the \u0027unique\u0027 filter to\nthe network_interfaces list, which does not guarantee a deterministic\nordering. This list is then transformed and passed to the\nstackhpc.libvirt-vm role.\n\nThere are two consequences of this:\n\n* it is not possible to determine which interface names should be used\n  prior to creating a seed VM\n* if a seed VM is recreated, the interface ordering may change\n\nThis change fixes the issue by sorting the network_interfaces list\nalphabetically before it is transformed and passed to the\nstackhpc.libvirt-vm.\n\nA new \u0027seed_vm_interfaces\u0027 variable is also added, which allows for\ncustomisation of the VM\u0027s interfaces - potentially allowing for more\ncomplex setups such as trunked VLANs.\n\nStory: 2007259\nTask: 38621\n\nThere is a second issue, which is that if the seed VM has a\nnetwork interface not configured with a gateway, cloud-init will fail to\nconfigure the network interfaces on the host. This has been observed on\nCentOS 8, but is probably more tied to the version of cloud-init, and\nmay affect CentOS 7. The following error is seen in the cloud-init logs:\n\n    KeyError: \u0027gateway\u0027\n\nThis change has been addressed in the jriguera.configdrive role, and\nthis change updates the version used in requirements.yml.\n\nStory: 2007769\nTask: 39993\n\nChange-Id: Ib6ab41a3ba320a1fe15d0d23561fad2fab7861e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c3b3e1826b28d6eac611e28378f5cdf815703f46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c3b3e1826b28d6eac611e28378f5cdf815703f46"}]},"branch":"refs/heads/master"},"b876d2e67e859d208234f132626e491303f5179b":{"kind":"REWORK","_number":4,"created":"2020-06-08 09:48:05.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/40/734040/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/734040/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/734040/4"}}},"commit":{"parents":[{"commit":"c2f049c6f01ee5d6dd383a5edef7fc051f79447f","subject":"Merge \"[Community goal] Update the contributor guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2f049c6f01ee5d6dd383a5edef7fc051f79447f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 08:53:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 09:48:02.000000000","tz":0},"subject":"Fix seed VM interface ordering","message":"Fix seed VM interface ordering\n\nCurrently the ordering of network interfaces in the seed VM is\nnon-deterministic. This happens because we apply the \u0027unique\u0027 filter to\nthe network_interfaces list, which does not guarantee a deterministic\nordering. This list is then transformed and passed to the\nstackhpc.libvirt-vm role.\n\nThere are two consequences of this:\n\n* it is not possible to determine which interface names should be used\n  prior to creating a seed VM\n* if a seed VM is recreated, the interface ordering may change\n\nThis change fixes the issue by sorting the network_interfaces list\nalphabetically before it is transformed and passed to the\nstackhpc.libvirt-vm.\n\nA new \u0027seed_vm_interfaces\u0027 variable is also added, which allows for\ncustomisation of the VM\u0027s interfaces - potentially allowing for more\ncomplex setups such as trunked VLANs.\n\nStory: 2007259\nTask: 38621\n\nThere is a second issue, which is that if the seed VM has a\nnetwork interface not configured with a gateway, cloud-init will fail to\nconfigure the network interfaces on the host. This has been observed on\nCentOS 8, but is probably more tied to the version of cloud-init, and\nmay affect CentOS 7. The following error is seen in the cloud-init logs:\n\n    KeyError: \u0027gateway\u0027\n\nThis change has been addressed in the jriguera.configdrive role, and\nthis change updates the version used in requirements.yml.\n\nStory: 2007769\nTask: 39993\n\nChange-Id: Ib6ab41a3ba320a1fe15d0d23561fad2fab7861e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b876d2e67e859d208234f132626e491303f5179b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b876d2e67e859d208234f132626e491303f5179b"}]},"branch":"refs/heads/master"},"7bcc5579e2de1735deaf970c5093a4ce629a37f0":{"kind":"REWORK","_number":5,"created":"2020-06-12 16:32:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/40/734040/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/40/734040/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/40/734040/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/40/734040/5"}}},"commit":{"parents":[{"commit":"c2f049c6f01ee5d6dd383a5edef7fc051f79447f","subject":"Merge \"[Community goal] Update the contributor guide\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/c2f049c6f01ee5d6dd383a5edef7fc051f79447f"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 08:53:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-12 16:32:48.000000000","tz":60},"subject":"Fix seed VM interface ordering","message":"Fix seed VM interface ordering\n\nCurrently the ordering of network interfaces in the seed VM is\nnon-deterministic. This happens because we apply the \u0027unique\u0027 filter to\nthe network_interfaces list, which does not guarantee a deterministic\nordering. This list is then transformed and passed to the\nstackhpc.libvirt-vm role.\n\nThere are two consequences of this:\n\n* it is not possible to determine which interface names should be used\n  prior to creating a seed VM\n* if a seed VM is recreated, the interface ordering may change\n\nThis change fixes the issue by sorting the network_interfaces list\nalphabetically before it is transformed and passed to the\nstackhpc.libvirt-vm.\n\nA new \u0027seed_vm_interfaces\u0027 variable is also added, which allows for\ncustomisation of the VM\u0027s interfaces - potentially allowing for more\ncomplex setups such as trunked VLANs.\n\nStory: 2007259\nTask: 38621\n\nThere is a second issue, which is that if the seed VM has a\nnetwork interface not configured with a gateway, cloud-init will fail to\nconfigure the network interfaces on the host. This has been observed on\nCentOS 8, but is probably more tied to the version of cloud-init, and\nmay affect CentOS 7. The following error is seen in the cloud-init logs:\n\n    KeyError: \u0027gateway\u0027\n\nThis change has been addressed in the jriguera.configdrive role, and\nthis change updates the version used in requirements.yml.\n\nStory: 2007769\nTask: 39993\n\nChange-Id: Ib6ab41a3ba320a1fe15d0d23561fad2fab7861e6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bcc5579e2de1735deaf970c5093a4ce629a37f0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7bcc5579e2de1735deaf970c5093a4ce629a37f0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
