)]}'
{"id":"openstack%2Fkayobe~734055","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I8e1d877ee91a07b86cb858d25b841f8bfcd50e21","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I8e1d877ee91a07b86cb858d25b841f8bfcd50e21","subject":"CentOS 8: separate kolla build tag from deploy tag","status":"MERGED","created":"2020-06-08 10:03:39.000000000","updated":"2020-06-19 12:28:57.000000000","submitted":"2020-06-19 12:26:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734055-1592569596804-69399470","meta_rev_id":"6deb8bc32c7df03b2e8995d88c8e4133153ec9cc","_number":734055,"virtual_id_number":734055,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 12:26:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-06-19 10:29:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-08 12:27:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-06-19 10:29:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"date":"2020-06-18 21:22:42.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-08 12:27:17.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-18 21:22:42.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-19 10:29:04.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-06-19 12:26:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"53e63840a82f015676a43684b4b9fa57ca4458bb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-08 10:03:39.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"4764a9bbda68a3f93c72bac27547a5f05a02e149","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-08 11:56:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61bcc961828946a9967b7bff7b765ac0 : SUCCESS in 5m 35s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2dbc04c30cec443a80891c5917e8f57a : SUCCESS in 6m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fdf273ba00224d6bacd9d1d2da264ab2 : SUCCESS in 6m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/158b449fdaac4366bc71367c89aca0d0 : SUCCESS in 6m 17s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/2287acd2db5542aaa7a3136eecfcb748 : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f85e0b7759a54c908cf9f24248b1836e : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/368b9ff6895b488f9629dee0742d441b : SUCCESS in 3m 30s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d85871526a0b4e7490e98fe8b917ff21 : SUCCESS in 4m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/434a29799b4b4ec4a976bdf0fffe4241 : SUCCESS in 9m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08e035fecda24698bc72683c3d918d29 : SUCCESS in 37m 27s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/5d88c62d62984417bb8e3ef3b5b09452 : SUCCESS in 1h 19m 49s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a818e41e0192467d926095996758f072 : SUCCESS in 56m 53s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/96ac9f94de1943c4955b19d5406228cd : SUCCESS in 16m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/58532b65662e46e69fab1a95c396605a : SUCCESS in 15m 49s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2c87ae9c1c4146e5995cd8d9e8f67262 : SUCCESS in 1h 47m 30s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/188d0e152b0746b58b15a7757cf8cdd2 : SUCCESS in 22m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e1ea5456edcd42e2881e5c5d4187feb3 : SUCCESS in 17m 40s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/2486e687d3864b6ebfb7e7c60c7bb84c : SUCCESS in 38m 57s","accounts_in_message":[],"_revision_number":1},{"id":"73068f629890191b9a06d48bd12c4ac7291e39ce","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-08 12:27:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"54993fae88546c82585fe74e168c73228c55af5c","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-19 10:29:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3771eaa444a9bb31f9dd1673c5cdd9a471d7f0c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 10:29:16.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"605267ce2ee529f4dea531609a39041a8b35b02d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:26:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9f2e4b8e81cd70db6bfa409bab751e0d50ed4d94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:26:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f0ae0bf682d54bacb96bd036f77b67b1 : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/87f5193bf949425a985447a040ee3c33 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6153f71c00dc42cfb5d54474659d14dd : SUCCESS in 4m 47s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b9afd7cc0b8f4bcf927e3d78a0933507 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f758ce80ccad428e87fb24844d28b184 : SUCCESS in 7m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f0004ae4bff04ec6b161fe615cc38a25 : SUCCESS in 3m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84c84fb45d2a4b679a86a2c55c2f18c1 : SUCCESS in 4m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5bcfd2318514435caf22fac15b9c6d17 : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9e13e12108b14251b5e7ba7c1bd173b6 : SUCCESS in 35m 39s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/315230a4d23a4c128e8fe6710cead519 : SUCCESS in 1h 15m 24s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e21151a2a2224ca3a845a717372cdf31 : SUCCESS in 1h 15m 02s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/5e85540baa634d14a97f200870076b88 : SUCCESS in 15m 20s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ef6e03542522449ebe8d6c9d7e71e9af : SUCCESS in 15m 25s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/0d98b693483f41e0a05ef6956c300934 : SUCCESS in 1h 56m 53s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/5c60e88a84b3444da92f03e4fe53c443 : SUCCESS in 19m 36s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/92ef2092b73c4b90981a263ee72e94ed : SUCCESS in 13m 54s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/64bc7487391e448ea212565bfb101d85 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"6deb8bc32c7df03b2e8995d88c8e4133153ec9cc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 12:28:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cd13b4da9aa44818cb1435507d115fd : SUCCESS in 1m 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f68f709343ba480eacbedca5842d2920 : SUCCESS in 1m 22s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33724452dd82e529a4852aa1bbdddafc829c12e7","revisions":{"33724452dd82e529a4852aa1bbdddafc829c12e7":{"kind":"REWORK","_number":1,"created":"2020-06-08 10:03:39.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/55/734055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/55/734055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/734055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/734055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/55/734055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/55/734055/1"}}},"commit":{"parents":[{"commit":"8e612e8954780fdd3c0835453b5ad7b57810b73a","subject":"Merge \"CI: mark pytest results as a Zuul artifact\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8e612e8954780fdd3c0835453b5ad7b57810b73a"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-03-10 17:33:32.000000000","tz":0},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 10:03:39.000000000","tz":0},"subject":"CentOS 8: separate kolla build tag from deploy tag","message":"CentOS 8: separate kolla build tag from deploy tag\n\nPrevously the container image tag applied to built images was configured\nvia \u0027kolla_openstack_release\u0027. This variable also controlled the tag\nused for deployed images. This could cause problems during the CentOS 8\ntransition, where we need to build two sets of images, and Kolla Ansible\nmay apply a \u0027-centos8\u0027 suffix to the tag we specify on CentOS 8.\n\nThis change separates the tag applied to built images into a different\nvariable - \u0027kolla_tag\u0027. The default is still \u0027kolla_openstack_release\u0027.\n\nChange-Id: I8e1d877ee91a07b86cb858d25b841f8bfcd50e21\nStory: 2006574\nTask: 39487\n(cherry picked from commit c563f45fb7e90dc202814df9f746bc909f2d0ffd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33724452dd82e529a4852aa1bbdddafc829c12e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/33724452dd82e529a4852aa1bbdddafc829c12e7"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
