)]}'
{"id":"openstack%2Fkayobe~734816","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I2b19e51b7b319b0243edab77065bba20820bebbf","project":"openstack/kayobe","branch":"stable/ussuri","topic":"create-ussuri","hashtags":[],"change_id":"I2b19e51b7b319b0243edab77065bba20820bebbf","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","status":"MERGED","created":"2020-06-10 12:48:55.000000000","updated":"2020-06-10 17:32:29.000000000","submitted":"2020-06-10 17:26:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734816-1591809997398-b1fdb8ea","meta_rev_id":"b185067b882a202b3553c2fd43fd4cc67ed58edd","_number":734816,"virtual_id_number":734816,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 17:26:37.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-10 15:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-06-10 15:24:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-06-10 15:24:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 15:23:24.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-10 15:24:54.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-10 17:26:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f9a3a61b2b2ecb97145c574df01eda29654a901d","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-06-10 12:48:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a702ddc3af91eff2607b3a7d627d3338953a4bbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 13:34:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4dfb4a73bca2411c8f5784833916d98f : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9929971512c4ad6ac5b89adee5e4f27 : SUCCESS in 6m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8224fe926eef4d21936a7d73a90df2c6 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ea58c6113c584f60a6d4f5b74081230c : SUCCESS in 8m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdee8804e2da41c4b7e732294a90bc92 : SUCCESS in 7m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49de3f4808434f69b51d237a2d12d6b9 : SUCCESS in 4m 35s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/49d21fab93914bb7bae14a071360dcf1 : SUCCESS in 6m 04s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f4bf6c71c081470c87a2fa7917235ab1 : SUCCESS in 9m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/2f4a028b3ee144a88f5ceca8b8de81df : SUCCESS in 36m 03s","accounts_in_message":[],"_revision_number":1},{"id":"dbe915685a5c03800157c2b3af18bf990fceb8ba","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-10 15:22:25.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"5bf3b1c0524b5bebad71172ee49b03d2a63ad6dc","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-10 15:23:24.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1300a962eaa15bde68462b3bdd9ab2ac892dd5c4","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-10 15:24:54.000000000","message":"Patch Set 1: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"e78fc840411e4fc056ad797679357d42b2a54470","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 15:26:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac227d5504f7bc8187ca3c8793f4f7b3bb1e3875","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 17:26:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3842e365063841bebebd1ec39a96e0ea : SUCCESS in 6m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bd435ddb070c40249c74a8c9797413e7 : SUCCESS in 8m 00s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3d22a201af77457c8b8a060824702b43 : SUCCESS in 7m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60e76bb8c3284065b16660a1f1ecdb03 : SUCCESS in 8m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ff264cd10a34a498eef93ff2263dd05 : SUCCESS in 3m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/33825152df814b9c8082129f20130308 : SUCCESS in 6m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d0d23c10875d49aa8c1608101fa0a09e : SUCCESS in 11m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/93e69bc4e84540e8a885184d1f29122d : SUCCESS in 30m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5d36244ccc098fa8a638da516650e40c92c3c0c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 17:26:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"df3aa6753727419a3322ad5c03d721a655214480","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 17:32:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9ff44fc1ef24acda10fd63b4c894a3a : SUCCESS in 2m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a88ee2a64bbb4b8fb34719aa0b27e778 : SUCCESS in 1m 29s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"38dd36f98c1064845a336b0c69338a98e05468f4","revisions":{"38dd36f98c1064845a336b0c69338a98e05468f4":{"kind":"REWORK","_number":1,"created":"2020-06-10 12:48:55.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/16/734816/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/16/734816/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/734816/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/734816/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/16/734816/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/16/734816/1"}}},"commit":{"parents":[{"commit":"2d8d85ea84da43e48bd5fc5795c7eb4491406edb","subject":"Update .gitreview for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/2d8d85ea84da43e48bd5fc5795c7eb4491406edb"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-10 12:48:52.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-10 12:48:52.000000000","tz":0},"subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","message":"Update TOX_CONSTRAINTS_FILE for stable/ussuri\n\nUpdate the URL to the upper-constraints file to point to the redirect\nrule on releases.openstack.org so that anyone working on this branch\nwill switch to the correct upper-constraints list automatically when\nthe requirements repository branches.\n\nUntil the requirements repository has as stable/ussuri branch, tests will\ncontinue to use the upper-constraints list on master.\n\nChange-Id: I2b19e51b7b319b0243edab77065bba20820bebbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38dd36f98c1064845a336b0c69338a98e05468f4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38dd36f98c1064845a336b0c69338a98e05468f4"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
