)]}'
{"id":"openstack%2Fkayobe~734818","triplet_id":"openstack%2Fkayobe~master~Ie2c272d1f185083537274adb534001b2d8664154","project":"openstack/kayobe","branch":"master","topic":"add-victoria-python-jobtemplates","hashtags":[],"change_id":"Ie2c272d1f185083537274adb534001b2d8664154","subject":"Add Python3 victoria unit tests","status":"MERGED","created":"2020-06-10 12:49:00.000000000","updated":"2020-06-10 17:33:25.000000000","submitted":"2020-06-10 17:26:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"734818-1591809999000-59896c92","meta_rev_id":"a4e33e8026e179db92407f33b5f564c25ea7a043","_number":734818,"virtual_id_number":734818,"owner":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-10 17:26:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-10 15:27:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-10 15:23:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-06-10 15:27:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-10 15:23:09.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-10 15:27:05.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-10 17:26:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"add0f9b84533f240fa5cb054bd85994ba9f5c7bf","author":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"date":"2020-06-10 12:49:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cb1266e7e03bf946e72c880674bdb2cb007f4709","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 15:00:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6174956b3a845068f178bccb2823a26 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9fa1adbd750468884a735887111c44d : SUCCESS in 6m 00s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c1427d3f510241c080a1666ba3d16cc8 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8d66bc1c6ac46ae9cae7e66960c30ff : SUCCESS in 6m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59cc7139829e42f2bfc0a2e3c946c572 : SUCCESS in 8m 14s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3c5b48bb2f6e4f70bfa69ac5a658e348 : SUCCESS in 5m 06s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e8b86c6664fa4e548055ec028ddeb898 : SUCCESS in 9m 59s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5ac644fada414c3383c8ac757643ff1c : SUCCESS in 35m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/2d8a2080365446c2894126f2d8d494d4 : SUCCESS in 1h 22m 30s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8f9055ef39484ba9bff8be3aa723c1f2 : SUCCESS in 16m 08s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ef924cdd6f57472fbe3dc5095c304670 : SUCCESS in 1h 59m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/95c00a879c1042e9b22679a9d7f2b7b8 : SUCCESS in 18m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2fd922c104a049738f328067a6af0375 : SUCCESS in 32m 36s","accounts_in_message":[],"_revision_number":1},{"id":"2025443c36541a9a33e57ee456793a1ff07c1092","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-10 15:23:09.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"a074de6ff6d99d185d2a0ca3bc0c577373101d27","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-10 15:27:05.000000000","message":"Patch Set 1: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64d5a201c0542fe3ca3d44669a383fbaee2da31b","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-10 15:27:41.000000000","message":"Patch Set 1: -Backport-Candidate Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0f608187ab423e1ab407e5922eb0920e6e19633b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 15:29:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7656ea8e5f2c1443c358e9bbd9f739cab600c202","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 17:26:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89a5c2ec97434fda8d73ad2ed10e64cf : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/723596ee4d724604ba6ce8c8b9ca695c : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d88a4cefe00f47149c4b00efffcb27fd : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6d602659ab54becb3704f11499d83fb : SUCCESS in 9m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6dd7096d89d14a38ac84c9ea5f663d12 : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ca4e2b17f0a84664a4f0b9c851cfddff : SUCCESS in 9m 55s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1fccade360f74f3a8c318adf5058c1c4 : SUCCESS in 26m 52s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/308f6c6472764561ab84ea119427bab4 : SUCCESS in 1h 17m 38s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4c3268d313e045268542679ec452e13c : SUCCESS in 16m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/89a3caaaa458484f94111a9733d8c898 : SUCCESS in 1h 43m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e3e9a2f35e4d4eb293c386c0387f0396 : SUCCESS in 17m 29s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3b1d7ed2f2c84b5c92e6029274a5a382 : SUCCESS in 29m 26s","accounts_in_message":[],"_revision_number":1},{"id":"820c7193db2fdcaf7acdf2dc19b900bc0820f05f","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 17:26:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"22c6a3173f903115b2b4fe264e62d40c64456c48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-10 17:33:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02114b42249e4278b67d6a5184a9a8b3 : SUCCESS in 1m 28s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b960d94b84e874cdcdacdacbf94e06670f3eb61","revisions":{"4b960d94b84e874cdcdacdacbf94e06670f3eb61":{"kind":"REWORK","_number":1,"created":"2020-06-10 12:49:00.000000000","uploader":{"_account_id":22816,"name":"OpenStack Release Bot","email":"infra-root@openstack.org","username":"release","tags":["SERVICE_USER"]},"ref":"refs/changes/18/734818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/18/734818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/734818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/734818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/18/734818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/18/734818/1"}}},"commit":{"parents":[{"commit":"34499fa56a9211ae5b42b641a3654e79ec3ab294","subject":"Update master for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/34499fa56a9211ae5b42b641a3654e79ec3ab294"}]}],"author":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-10 12:48:58.000000000","tz":0},"committer":{"name":"OpenStack Release Bot","email":"infra-root@openstack.org","date":"2020-06-10 12:48:58.000000000","tz":0},"subject":"Add Python3 victoria unit tests","message":"Add Python3 victoria unit tests\n\nThis is an automatically generated patch to ensure unit testing\nis in place for all the of the tested runtimes for victoria.\n\nSee also the PTI in governance [1].\n\n[1]: https://governance.openstack.org/tc/reference/project-testing-interface.html\n\nChange-Id: Ie2c272d1f185083537274adb534001b2d8664154\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b960d94b84e874cdcdacdacbf94e06670f3eb61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4b960d94b84e874cdcdacdacbf94e06670f3eb61"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
