)]}'
{"id":"openstack%2Fkayobe~735107","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I2b376303a9fef8fb3c1ada0c168bbb494216aa94","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I2b376303a9fef8fb3c1ada0c168bbb494216aa94","subject":"Disable broken linting of literalinclude sphinx directive","status":"MERGED","created":"2020-06-11 10:54:58.000000000","updated":"2020-06-11 13:10:55.000000000","submitted":"2020-06-11 13:08:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735107-1591880909454-cef858b0","meta_rev_id":"18f38bba1ad5d66aa7545293bcfd8f98312a5754","_number":735107,"virtual_id_number":735107,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-11 13:08:29.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-11 12:29:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-06-11 12:01:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-06-11 12:29:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 12:29:59.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-11 13:08:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"72b316c261b6be3b6dc41809ee5c654deb905224","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-11 10:54:58.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"29a9fc18f387d296183f1e1973506cba72bccbdb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 11:32:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/82f99376c7ac4a51a1d011efd77ae975 : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ac787134ea34814b3cb29f21a7a32a7 : SUCCESS in 5m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6552b9622b7f438cae73d916aad95035 : SUCCESS in 4m 55s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c630ab04be364ab89ab6ea0b8b56da97 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb90a23cfcbe4d3c86b5ceebf3ae5a25 : SUCCESS in 7m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/de7c2f4a6ce14104a14ddfda8cfe76e2 : SUCCESS in 4m 52s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b93f07a8a4a64437ae1420c156882f18 : SUCCESS in 9m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f7c2d117edb04797bfa525d87d858bcf : SUCCESS in 36m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c4d390722999ff3685965c29dbbc5fa99e97e17d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-11 12:01:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"280e63116d41c11e0a171be00445201348389aff","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-11 12:29:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"51f8e32ca9a99f63d4910b2bc1fd6dde7054d8d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 12:30:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3c67e1cae26480a3bedf9634b175d72f0dfc796a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 13:08:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c8979f221db0d55fa8469f71a41e4ee25813441c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 13:08:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22bedf2e7254360876860739a8d8e3c : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/48b4a9a41c94477f8a82e15678f8c7a2 : SUCCESS in 5m 15s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c59f65d32f3c435c876a2e260e0fea9f : SUCCESS in 5m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/776c49073132450183f30e5b8d837a8e : SUCCESS in 8m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c2699cf240747ddb46e8cd5236173d8 : SUCCESS in 6m 03s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a4d1e2cfeac34c9495712b3fbc43808a : SUCCESS in 8m 49s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6c42aec2ec1a4622b21be3ebf274d4c9 : SUCCESS in 33m 23s","accounts_in_message":[],"_revision_number":1},{"id":"18f38bba1ad5d66aa7545293bcfd8f98312a5754","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 13:10:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93d64d2306d14649a1ff43e045941085 : SUCCESS in 1m 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a3c3e0b90451118509709a64312fb29637144786","revisions":{"a3c3e0b90451118509709a64312fb29637144786":{"kind":"REWORK","_number":1,"created":"2020-06-11 10:54:58.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/07/735107/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/07/735107/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/735107/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/735107/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/07/735107/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/07/735107/1"}}},"commit":{"parents":[{"commit":"38dd36f98c1064845a336b0c69338a98e05468f4","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38dd36f98c1064845a336b0c69338a98e05468f4"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-06-10 18:27:25.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 10:54:58.000000000","tz":0},"subject":"Disable broken linting of literalinclude sphinx directive","message":"Disable broken linting of literalinclude sphinx directive\n\nPreviously we were seeing:\n\nAttributeError: \u0027Values\u0027 object has no attribute \u0027env\u0027\n\nTrivialFix\nChange-Id: I2b376303a9fef8fb3c1ada0c168bbb494216aa94\n(cherry picked from commit c96803d8d1974c87e0dca004b4e360d48ee0fe8f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3c3e0b90451118509709a64312fb29637144786"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3c3e0b90451118509709a64312fb29637144786"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
