)]}'
{"id":"openstack%2Fkayobe~735108","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I6f065073feee42b2a0921a4aea98f7004fe5f7bf","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I6f065073feee42b2a0921a4aea98f7004fe5f7bf","subject":"Add framework to deploy user-defined containers on seed","status":"MERGED","created":"2020-06-11 10:55:16.000000000","updated":"2020-06-17 19:16:11.000000000","submitted":"2020-06-17 19:09:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735108-1592420979624-8bc90565","meta_rev_id":"c82c3194854a3cb4ab2612845a209d4230403a4c","_number":735108,"virtual_id_number":735108,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 19:09:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-06-17 15:29:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":2,"date":"2020-06-16 10:09:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-06-17 15:28:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-06-17 15:29:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":1,"date":"2020-06-17 15:28:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"value":1,"date":"2020-06-16 10:09:41.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-11 10:55:16.000000000","updated_by":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"reviewer":{"_account_id":22629,"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","username":"mnasiadka"},"state":"REVIEWER"},{"updated":"2020-06-17 15:28:02.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-17 15:29:02.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-06-17 19:09:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6450aff74df5ca0026f3349686715536f35331b1","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-11 10:55:16.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"91ff35462700584ab203dc0966ec2ab793134d3b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-11 10:55:27.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"78a1168a72acf7d56d05e39ed3f4806e0e929d1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 12:39:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/495a4c0b1fc94483b03eeadc68479334 : SUCCESS in 4m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69492eb5f11d4c67b580f12e4478029d : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aa938329890c4951975a16c4898eac72 : SUCCESS in 4m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/c37de89b62a844f89054f4c7ad0dd983 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793a3532e2e74000aca3d399468bc885 : SUCCESS in 6m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/62cedfca96b94c4b93109ce5a7a69038 : SUCCESS in 3m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3846ea3a65dd45aca8285621848542ac : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c1571e67d5f4d419392e1c587eab254 : SUCCESS in 8m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d4f905f8fe4346128ba38756b0e3de0d : SUCCESS in 27m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d6fcd4e91502469ba6c7cb537d0a99e2 : SUCCESS in 1h 10m 11s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/efd496283aaf4053bc8717254b7c6ac5 : SUCCESS in 15m 54s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2437f86747244f05acb9afad42f2e8f1 : SUCCESS in 1h 39m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2f79713d390b4a8d9136a81ea0883726 : SUCCESS in 15m 09s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/34b0a9ff9b3446e9ae9aa9d65a469d17 : SUCCESS in 26m 34s","accounts_in_message":[],"_revision_number":2},{"id":"859020022de86e15dadad5561a4b3e6686c5e200","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 10:09:37.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"ec0018ecb145527417dd2b32d1393f533ea297e6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 10:09:41.000000000","message":"Patch Set 2: Review-Priority+1","accounts_in_message":[],"_revision_number":2},{"id":"c69a217deb4c1373d61e2534fea7b4ac75781c2e","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-17 15:28:02.000000000","message":"Patch Set 2: Backport-Candidate+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a3a81c84e148fab4b5b2acfbbaf6d7021ed2f215","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 15:28:16.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"caf12fab3013542a5fc87c567b72db10090590c8","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-17 15:29:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"01e7cb028b0afe82c3d8a2a2159dad26b02967dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:09:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"930e2b398544b2bbc74ca83128029112775358cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:09:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5455d55412664ca091853ea00d4e07ce : SUCCESS in 5m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b2e2f74921c45e4908fe10c5959a38b : SUCCESS in 5m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0a1fc46028b04cb2abc7b53448b3de7a : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0582e10feb4ac4bdffad5bf21f169e : SUCCESS in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14f36ec3cfca453eb87f29f311d601f6 : SUCCESS in 6m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a7ddc51db05a44a5b898935fe64b426b : SUCCESS in 9m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b8eabafa418042ed8509ad0c37a9da00 : SUCCESS in 8m 51s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/aec3c223a48a42b39a1d532fb2b18ff2 : SUCCESS in 35m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/688223d5b0944c3688c23779dccc04f8 : SUCCESS in 1h 11m 28s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e485dce3ed2b458b9afcc3c97ee0587c : SUCCESS in 18m 58s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7631b52978524b7299557c191c5adb61 : SUCCESS in 1h 52m 56s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ecf9a6954ef543f2bb00fd5b557987db : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b6fa727ba5174235ad896a4d729a283e : SUCCESS in 29m 02s","accounts_in_message":[],"_revision_number":2},{"id":"c82c3194854a3cb4ab2612845a209d4230403a4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:16:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6237ffdab724bb5bfc50c626f17c4e5 : SUCCESS in 1m 32s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/14e0b207b45c4b3e948d18f3e858ad09 : SUCCESS in 1m 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"daa7fd5e65bf0159559ea3763ca61bc0b2777df6","revisions":{"af8fcdb9f9a5a132f5d786d979c6758dbe28cf8e":{"kind":"REWORK","_number":1,"created":"2020-06-11 10:55:16.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/735108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/08/735108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/735108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/735108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/735108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/08/735108/1"}}},"commit":{"parents":[{"commit":"38dd36f98c1064845a336b0c69338a98e05468f4","subject":"Update TOX_CONSTRAINTS_FILE for stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/38dd36f98c1064845a336b0c69338a98e05468f4"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 10:55:16.000000000","tz":0},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nTask: 39811\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n(cherry picked from commit aefd6e5a56ba9b172a8111f09c621afd09041a47)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af8fcdb9f9a5a132f5d786d979c6758dbe28cf8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/af8fcdb9f9a5a132f5d786d979c6758dbe28cf8e"}]},"branch":"refs/heads/stable/ussuri"},"daa7fd5e65bf0159559ea3763ca61bc0b2777df6":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-11 10:55:27.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/08/735108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/08/735108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/735108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/735108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/08/735108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/08/735108/2"}}},"commit":{"parents":[{"commit":"a3c3e0b90451118509709a64312fb29637144786","subject":"Disable broken linting of literalinclude sphinx directive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3c3e0b90451118509709a64312fb29637144786"}]}],"author":{"name":"Michal Nasiadka","email":"mnasiadka@gmail.com","date":"2020-02-25 10:31:08.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 10:55:27.000000000","tz":0},"subject":"Add framework to deploy user-defined containers on seed","message":"Add framework to deploy user-defined containers on seed\n\nTask: 39811\n\nChange-Id: I6f065073feee42b2a0921a4aea98f7004fe5f7bf\n(cherry picked from commit aefd6e5a56ba9b172a8111f09c621afd09041a47)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/daa7fd5e65bf0159559ea3763ca61bc0b2777df6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/daa7fd5e65bf0159559ea3763ca61bc0b2777df6"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
