)]}'
{"id":"openstack%2Fkayobe~735148","triplet_id":"openstack%2Fkayobe~master~Ideada0f299f34d16d1286633b6e21404b736918b","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"Ideada0f299f34d16d1286633b6e21404b736918b","subject":"Fix ironic inspector store endpoint configuration","status":"MERGED","created":"2020-06-11 14:25:29.000000000","updated":"2020-06-18 07:51:18.000000000","submitted":"2020-06-17 19:02:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735148-1592420576548-2c7c96e1","meta_rev_id":"0f9d8f2474ce5f725b97941b63fb8d8d7167976b","_number":735148,"virtual_id_number":735148,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 19:02:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-06-17 14:55:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-15 20:38:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-06-17 14:55:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-15 20:38:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 20:38:48.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-17 14:55:07.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-06-17 19:02:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bf742256346df6abf484ef2f99f8997c7cf68d8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-11 14:25:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d7030c0b9b82a39f342e392cfdd1deec36e339a9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-11 14:28:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"37992c5028855b47a353ffc19f46278cbb242f93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 16:34:33.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb813a903d62407ab439abec42199c71 : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8064119966d42c2b47742b1e84911d0 : SUCCESS in 4m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/766a468fd17149bda17908598839a2e4 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/067e46ecbb464b62909ab8650398044a : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c2edfd0869c43f396054dd9e27512db : SUCCESS in 5m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a9d14105f8745358ba48030a37774c7 : SUCCESS in 4m 16s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e31e827d4a0248f9a0eece14073ced7e : SUCCESS in 8m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/013e5022932141b8b342b7a80688fbb8 : SUCCESS in 35m 45s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/208de15df845445fab069c90dfe64b46 : SUCCESS in 1h 09m 08s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/998883602e704c6380fc1b81fd669c42 : SUCCESS in 17m 14s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/81bb0a13169b4d6c8f3c57dc6f0e1d34 : SUCCESS in 2h 01m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/b676462406b24fe1815a54a247c65a6b : SUCCESS in 17m 18s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ef0b6e8c040c481eb1fc0d7a68493360 : SUCCESS in 30m 43s","accounts_in_message":[],"_revision_number":2},{"id":"c3154177274d2d985fac1c050093ef75b7ca1d6c","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-11 16:48:10.000000000","message":"Patch Set 3: Published edit on patch set 2.","accounts_in_message":[],"_revision_number":3},{"id":"eee6bbad442d6273acbce836c937569a3f700594","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 18:52:08.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba7a782e38cf4e178859ac40cf9d6022 : SUCCESS in 6m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6988fc95858d4018b0f5ef26ba8042f0 : SUCCESS in 4m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1fc01c5f48c646bdbea0bab3413876bd : SUCCESS in 6m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1f186f22fa7d42f6bd5ef4e541ecece9 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7c4072beb554e879e976ec479f7b207 : SUCCESS in 6m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/1ed32838e644435fa2432f9d09f246eb : SUCCESS in 5m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/bc740588251c415b9e743b46ba8be266 : SUCCESS in 8m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/a51cdb7b49d24c8fbedffdbd5a0cc095 : SUCCESS in 33m 16s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4672d903236f42ce9107090ea10e1968 : SUCCESS in 1h 07m 21s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a87415f8628a448f91a05f99eb29ef6f : SUCCESS in 16m 34s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6ef5d2a102374827a7e769feff8182ab : SUCCESS in 1h 51m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/52fa5a58a17c4fb88bd03a31d27a273b : SUCCESS in 16m 37s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/51dc53d107ae451e9fcfd06abde68b2f : SUCCESS in 27m 41s","accounts_in_message":[],"_revision_number":3},{"id":"f4854a28ffe7419c9f4c9bffac882c199645c7f9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-15 20:38:48.000000000","message":"Patch Set 3: Backport-Candidate+2 Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0e04c27a3bcd575114a0ab1d12c4a528457a4c5d","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-17 14:55:07.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a1b7f1d71eda63f92a6654004ffb92e8e1ef3d7f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 14:55:23.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"80588d75c222888da1d5df433199aa75115b8315","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:02:56.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f605b90c42b440d817ccdf152ea60b6 : SUCCESS in 7m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f12d8fbbd8584a09a3bb70aa8b0e772f : SUCCESS in 7m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/187bbcf7a350465581205bd75f9e4e8e : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae556185cf0f40b0ab06663952638ac8 : SUCCESS in 9m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b0d0dbf5c96f4b6c934631710f3da888 : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/44aea0fbbda244b5a51ffa704fff573c : SUCCESS in 8m 11s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/96376481ab8e4e9b89a8d94afb0e33f5 : SUCCESS in 37m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8e32769d435a4c8c84b2c3142d69f0bd : SUCCESS in 1h 15m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e108198516514014ad266486417bd2c2 : SUCCESS in 17m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8e0d0fd3f9a34541865a920a9135d61c : SUCCESS in 1h 57m 12s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/4f170098f44d4d2da1969be16f5297f0 : SUCCESS in 19m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/683fb3983095481f952c11789a13ecea : SUCCESS in 25m 52s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/4df3860a16274bb99dd7a4f388e0244b : SUCCESS in 14m 41s","accounts_in_message":[],"_revision_number":3},{"id":"d77d9302766a69a051de2011aa2c573e703c75af","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:02:56.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"3fdcd271bb44970d8043951f9b76bfb2af9594a5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:06:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33916e7604b5438eac37c8ff3d8ca344 : SUCCESS in 2m 21s","accounts_in_message":[],"_revision_number":3},{"id":"5e3c7820d66072877f309cbca769ade3481d0374","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:51:04.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 9ba8413814342afdb436f62aed4912a32d9bc0c7","accounts_in_message":[],"_revision_number":3},{"id":"0f9d8f2474ce5f725b97941b63fb8d8d7167976b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:51:18.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 31b90d5199b7988b3b98b0f1cc6340b47b7084e3","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"f0d1b630fc960923717fb0b3e85f56935298fd4d","revisions":{"340a7368ac5d7a82ca0dce63d2bb284b9c7a8d9d":{"kind":"REWORK","_number":1,"created":"2020-06-11 14:25:29.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/735148/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/48/735148/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/735148/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/735148/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/735148/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/48/735148/1"}}},"commit":{"parents":[{"commit":"e75b4a78546ca5cb9816ae2753f6863236f1d982","subject":"Correct a typo in the document","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e75b4a78546ca5cb9816ae2753f6863236f1d982"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 14:12:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 14:24:58.000000000","tz":60},"subject":"Fix ironic inspector store endpoint configuration","message":"Fix ironic inspector store endpoint configuration\n\nWhen performing inspection on baremetal compute nodes, inspector\ncurrently has the following failure when storing inspection data to the\nintrospection store:\n\n    No \u0027keystoneauth1.plugin\u0027 driver found, looking for \u0027token_endpoint\u0027\n\nThis happens because the token_endpoint auth type is no longer\navailable. The new way to do this is to set auth_type \u003d none and provide\nan endpoint override.\n\nStory: 2007795\nTask: 40049\n\nChange-Id: Ideada0f299f34d16d1286633b6e21404b736918b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/340a7368ac5d7a82ca0dce63d2bb284b9c7a8d9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/340a7368ac5d7a82ca0dce63d2bb284b9c7a8d9d"}]},"branch":"refs/heads/master"},"926da6e5c95ff1c2aaf4425215a88f95b98c3c3e":{"kind":"REWORK","_number":2,"created":"2020-06-11 14:28:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/735148/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/48/735148/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/735148/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/735148/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/735148/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/48/735148/2"}}},"commit":{"parents":[{"commit":"e75b4a78546ca5cb9816ae2753f6863236f1d982","subject":"Correct a typo in the document","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e75b4a78546ca5cb9816ae2753f6863236f1d982"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 14:12:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 14:27:44.000000000","tz":60},"subject":"Fix ironic inspector store endpoint configuration","message":"Fix ironic inspector store endpoint configuration\n\nWhen performing inspection on baremetal compute nodes, inspector\ncurrently has the following failure when storing inspection data to the\nintrospection store:\n\n    No \u0027keystoneauth1.plugin\u0027 driver found, looking for \u0027token_endpoint\u0027\n\nThis happens because the token_endpoint auth type is no longer\navailable. The new way to do this is to set auth \u003d none and provide\nan endpoint override.\n\nStory: 2007795\nTask: 40049\n\nChange-Id: Ideada0f299f34d16d1286633b6e21404b736918b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/926da6e5c95ff1c2aaf4425215a88f95b98c3c3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/926da6e5c95ff1c2aaf4425215a88f95b98c3c3e"}]},"branch":"refs/heads/master"},"f0d1b630fc960923717fb0b3e85f56935298fd4d":{"kind":"REWORK","_number":3,"created":"2020-06-11 16:48:10.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/48/735148/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/48/735148/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/735148/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/735148/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/48/735148/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/48/735148/3"}}},"commit":{"parents":[{"commit":"e75b4a78546ca5cb9816ae2753f6863236f1d982","subject":"Correct a typo in the document","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e75b4a78546ca5cb9816ae2753f6863236f1d982"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 14:12:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 16:48:06.000000000","tz":0},"subject":"Fix ironic inspector store endpoint configuration","message":"Fix ironic inspector store endpoint configuration\n\nWhen performing inspection on baremetal compute nodes, inspector\ncurrently has the following failure when storing inspection data to the\nintrospection store:\n\n    No \u0027keystoneauth1.plugin\u0027 driver found, looking for \u0027token_endpoint\u0027\n\nThis happens because the token_endpoint auth type is no longer\navailable. The new way to do this is to set auth_type \u003d none and provide\nan endpoint override.\n\nStory: 2007795\nTask: 40049\n\nChange-Id: Ideada0f299f34d16d1286633b6e21404b736918b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0d1b630fc960923717fb0b3e85f56935298fd4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/f0d1b630fc960923717fb0b3e85f56935298fd4d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
