)]}'
{"id":"openstack%2Fkayobe~735182","triplet_id":"openstack%2Fkayobe~master~I4a72d9ae49ad41716522c3074c16d8ca23c3ff94","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I4a72d9ae49ad41716522c3074c16d8ca23c3ff94","subject":"CI: Update IPA images during upgrade","status":"MERGED","created":"2020-06-11 15:47:08.000000000","updated":"2020-06-18 07:52:36.000000000","submitted":"2020-06-17 19:08:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735182-1592420921391-34ccc9b0","meta_rev_id":"3c6222464cadd55db15462aa5c84a932fab922a2","_number":735182,"virtual_id_number":735182,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-17 19:08:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-06-17 15:00:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-15 20:52:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":1,"date":"2020-06-17 15:00:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 20:52:51.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-17 15:00:05.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-06-17 19:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"01a7dadea6b4c73897908f98f2b7967a7a5a0ed5","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-11 15:47:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"44a06d27d0810e287e95d9bcaf3c3716c2632e0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-11 17:14:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/adf5868ae031469fb2e81a80ce9b26a4 : SUCCESS in 10m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c73dc3db675a427eb11858ea11c1db7c : SUCCESS in 9m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b8e8e5a318e44a9d8473ab06b22cc013 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93855b2e295b4eb2bc6a28b9bf323483 : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/245b2e53812840ac978d405e66d2faf8 : SUCCESS in 9m 21s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/8c1f28783d834918b6607994d36726c0 : SUCCESS in 7m 13s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f03336920d044f2b559f37c3c094060 : SUCCESS in 12m 19s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/266f61ec8df745e7aa4744d0de14ff00 : SUCCESS in 35m 53s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6b8da8d62144405aa1186becaef6dc7e : SUCCESS in 1h 08m 47s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/791369290c18448aa742ffdb78f294e0 : SUCCESS in 18m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7d6576dcd8844dc6835465ca6b4a0177 : FAILURE in 1h 11m 31s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/de2bbe92706d4bffa7faa1cd3012d232 : SUCCESS in 20m 21s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/df8273314a4845c6ac5951369e48ba1a : SUCCESS in 28m 49s","accounts_in_message":[],"_revision_number":1},{"id":"5228e7ae471c39b8ad0bf2907b593d80765e05e4","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-12 16:01:43.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"2f61655e7631a2a231c77caaaabcddcf432424ff","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-12 16:25:11.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"8b8e14cd374eaaf23c1d8cf9036893f3b36a539e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 18:21:38.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7faf79de92d345679075089184a91495 : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a12cc3aff5c4f14bc85c11147fe9a97 : SUCCESS in 4m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b9fb2e262d440d796ab958b6b2e8e89 : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d151509975354c12867c5eff3ad921d0 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81bfa079e3bb422590bba1131eb4175c : SUCCESS in 5m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e4e13371a8e049cc9e7130f1974acb46 : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/50a8a1560fff45258da3adaa302bdd60 : SUCCESS in 9m 27s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0837f19af4214df89a3a922a76b4b2a8 : SUCCESS in 34m 21s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ac8c9918240d4a2fb4924ea1e34360e8 : SUCCESS in 1h 08m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/172a4540f44944a180bb896ee04227d7 : SUCCESS in 16m 04s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f076ecd18f4743699d44956f4ccc3e83 : SUCCESS in 1h 51m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/65d4c8cc1d82411daa528aec003993fc : SUCCESS in 13m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/db5dc00de4644a1c89f4632c2ecbf0b2 : SUCCESS in 27m 55s","accounts_in_message":[],"_revision_number":3},{"id":"00e7b61d2488e01e5b81671c8c0723b0ba8b9229","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-15 20:52:51.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"df85e20f84d2618f493846be4d0e9ce27aec952b","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-17 15:00:05.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a13221f982254252cf7fb3efc4d7714960d46884","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 15:00:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"02adbe3d830d2cb5e5afa978ac802a4ce0a7aa11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:08:40.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d58a0d702f24dabb17528550478ba33 : SUCCESS in 7m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bf6bff4e533f494fa5220ed4ac2fca8d : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e34c8fda6f54112b0a2c12c63cab937 : SUCCESS in 8m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fb67a5c6b8449e9841b70825b4942cc : SUCCESS in 9m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d0d8fd940ec74c4daf65e3dfeb203f03 : SUCCESS in 7m 39s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fdd27d7b9fed407094b29b7f34b69155 : SUCCESS in 8m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1d9e036126f3450c946a51bbb47b2159 : SUCCESS in 38m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/57cd62a977894756af7343034782cd93 : SUCCESS in 1h 19m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/70587ee0c4cc4661b8de8983a6789724 : SUCCESS in 20m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b394fbe9330344fd9ff83c259f2000b0 : SUCCESS in 2h 06m 14s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/20227c2a4f804806a797d3344bda93eb : SUCCESS in 20m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d294fdad34324c90ae90ea443907c30e : SUCCESS in 31m 58s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e8f86e05cbe04d569fec9a666c28a67a : SUCCESS in 14m 51s","accounts_in_message":[],"_revision_number":3},{"id":"fd2c74b5e952426c2316072c38acba668d4f7d23","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 19:08:41.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"fd414233610473b7e562d69ead87077b6f598a61","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:52:07.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/ussuri as commit 6e5570816102dbfd941f0db51db72e4beb2953c1","accounts_in_message":[],"_revision_number":3},{"id":"112868d5b1daa6e225d448ab456bf3bb1cfed396","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:52:25.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/train as commit 1774d9e92e3f91127e5c974a6a85d8802a0e2e79","accounts_in_message":[],"_revision_number":3},{"id":"3c6222464cadd55db15462aa5c84a932fab922a2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:52:36.000000000","message":"Patch Set 3: Cherry Picked\n\nThis patchset was cherry picked to branch stable/stein as commit 044dd31a0fccdc685b7386088e40aec1d4292eae","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"3d9c5861346d1d14e623b302bc6f5aacfa38c12e","revisions":{"e1a87de949decbcbcd26b3a75b0c320345d1acef":{"kind":"REWORK","_number":1,"created":"2020-06-11 15:47:08.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/735182/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/735182/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/735182/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/735182/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/735182/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/735182/1"}}},"commit":{"parents":[{"commit":"b5e5553526b45ca968ea29e29faf79cd7d36dd2d","subject":"IPA: Switch to IPA builder and CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5e5553526b45ca968ea29e29faf79cd7d36dd2d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 15:46:48.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 15:46:50.000000000","tz":60},"subject":"CI: Update IPA images during upgrade","message":"CI: Update IPA images during upgrade\n\nThis ensures we are using the appropriate IPA images in the upgraded\nenvironment.\n\nChange-Id: I4a72d9ae49ad41716522c3074c16d8ca23c3ff94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1a87de949decbcbcd26b3a75b0c320345d1acef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e1a87de949decbcbcd26b3a75b0c320345d1acef"}]},"branch":"refs/heads/master"},"db78c8ccd22503069f9b45b829c34de77cfd4553":{"kind":"REWORK","_number":2,"created":"2020-06-12 16:01:43.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/735182/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/735182/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/735182/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/735182/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/735182/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/735182/2"}}},"commit":{"parents":[{"commit":"b5e5553526b45ca968ea29e29faf79cd7d36dd2d","subject":"IPA: Switch to IPA builder and CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b5e5553526b45ca968ea29e29faf79cd7d36dd2d"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 15:46:48.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-12 16:01:41.000000000","tz":0},"subject":"CI: Update IPA images during upgrade","message":"CI: Update IPA images during upgrade\n\nThis ensures we are using the appropriate IPA images in the upgraded\nenvironment.\n\nChange-Id: I4a72d9ae49ad41716522c3074c16d8ca23c3ff94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db78c8ccd22503069f9b45b829c34de77cfd4553"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/db78c8ccd22503069f9b45b829c34de77cfd4553"}]},"branch":"refs/heads/master"},"3d9c5861346d1d14e623b302bc6f5aacfa38c12e":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-06-12 16:25:11.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/735182/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/735182/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/735182/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/735182/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/735182/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/735182/3"}}},"commit":{"parents":[{"commit":"20fb05bfb47b747679ead96a821952c11b940fc4","subject":"IPA: Switch to IPA builder and CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/20fb05bfb47b747679ead96a821952c11b940fc4"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 15:46:48.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-12 16:25:11.000000000","tz":0},"subject":"CI: Update IPA images during upgrade","message":"CI: Update IPA images during upgrade\n\nThis ensures we are using the appropriate IPA images in the upgraded\nenvironment.\n\nChange-Id: I4a72d9ae49ad41716522c3074c16d8ca23c3ff94\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d9c5861346d1d14e623b302bc6f5aacfa38c12e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d9c5861346d1d14e623b302bc6f5aacfa38c12e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
