)]}'
{"id":"openstack%2Fkayobe~735333","triplet_id":"openstack%2Fkayobe~master~I16c86ce90d2b5a689e69681f12d0dbcb5c4cd6be","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I16c86ce90d2b5a689e69681f12d0dbcb5c4cd6be","subject":"CI: set previous_release to ussuri","status":"MERGED","created":"2020-06-12 12:51:34.000000000","updated":"2020-06-16 12:54:29.000000000","submitted":"2020-06-16 12:50:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735333-1592311807424-03b49476","meta_rev_id":"5fadad574a6a97e217ea24fd342d5a6f532d3f9b","_number":735333,"virtual_id_number":735333,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 12:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-12 15:54:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-16 10:12:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-16 10:12:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-16 08:40:47.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-16 08:40:47.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-16 10:12:23.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-16 12:50:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a607e8f078db80be0ced0b75ebb12ecdf45f74c","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-12 12:51:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3678760909c5d814bf3e9393a8526ac2c16f2b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 14:39:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ce7cc3a853bd4bff82ecdc99ba851c75 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a6631797cae4e538206782a059db6db : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/579b1f3805ba4c7684da5a0b7577d35c : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae0ecc1041454e5a855567a082d52c17 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ceebca83b1c949f1a7e20086e717f5b0 : SUCCESS in 11m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2cea7d2b2dab4602b10c78141be5d02b : SUCCESS in 4m 36s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/64ea40ab48ab421fad73d01f570589a3 : SUCCESS in 10m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9afc805426794fe68e4c136d483da11c : SUCCESS in 31m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a1af6ed2592c48409a90f2f0e131480c : FAILURE in 32m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/102df4d653dc4bd9be16f457bcae32b5 : SUCCESS in 18m 51s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/33ca87ff838343e59b1f0b74a234905a : SUCCESS in 1h 46m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/3ff62f9d298e4e0bbd155aee8f329574 : SUCCESS in 15m 46s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e11218ef7378409e99d019991867e2c3 : SUCCESS in 25m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7a7b9a0e4a5da2985a0471fc81a01a875a862999","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-12 14:58:02.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"7bb6e6443cfba98b2146c6a0a28e1cc8218b0d45","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-12 15:54:14.000000000","message":"Patch Set 1: Code-Review+2\n\nFine if it passes","accounts_in_message":[],"_revision_number":1},{"id":"3baae59badce8ac423f943a6bdd18d864b305843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 17:05:07.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e9d864979f34e7288640c3af2934323 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d57f97562ef14a0590aec4a4a7265bf6 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/73b091acb4c84ffb903f52b43ec698db : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/521e69f7693b4461a4cc2784877e8133 : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad08292426624eb7a2a2a86c66857f90 : SUCCESS in 8m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/f093695c66ab470184e73974e2ce3343 : SUCCESS in 4m 53s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7b8a8f3b3d6149e48f59428493a3c39d : SUCCESS in 9m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/08fe3fe407504b80a9e351309ec40bd4 : SUCCESS in 33m 43s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/730de6e9e3a64afdad5e578146b18f50 : TIMED_OUT in 2h 05m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4793e62bb269401aaa2a341183526ac4 : SUCCESS in 12m 11s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cb8f9c31fc62415a92a53e73aa451025 : SUCCESS in 1h 40m 43s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5683453180fb4cf5abc4c26465878467 : SUCCESS in 16m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6f2573050e094b5caad2a5d105b2f60e : SUCCESS in 22m 50s","accounts_in_message":[],"_revision_number":1},{"id":"28e44e58a564dde5dff34aa2e16f03af52b1607e","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-12 19:27:54.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"fdf77d9e79acb7918eb43e154f7c8cedb0ccd5ef","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-15 16:07:25.000000000","message":"Patch Set 1:\n\nrecheck\n\nZuul where are you?","accounts_in_message":[],"_revision_number":1},{"id":"c12466811b60d88632f31c9f3ff0d2532abe0f86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 18:12:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1fffb19db0f44369855352a5cdd6aba3 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43d30db949194491a9240169efe3cb86 : SUCCESS in 7m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ce17c3ae3594c9b8edce7ad7bdf8e8f : SUCCESS in 8m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb1283e2f008408bb3014ce9167a7cf6 : SUCCESS in 7m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d13d009cd3e4c3db6655aca430a64d7 : SUCCESS in 9m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e8b71ba99ff84002a625ab3e76555b45 : SUCCESS in 7m 17s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f774d6d242064427a4a65c6cbc756b08 : SUCCESS in 11m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/783bef7ac7184dd68fa2d2cfc155f876 : SUCCESS in 31m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/560600e86392451e9af97aaec03cf16b : SUCCESS in 1h 19m 10s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/b42d1bc5014146569de5280b7a78ed26 : SUCCESS in 18m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/c30fa66d10ca4015922e9b3bf1fbf826 : SUCCESS in 1h 42m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/41c19b867fee4c7bb67eec1bcb08382c : SUCCESS in 22m 00s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8b387f3b9eac4b46907f22f996d12fff : SUCCESS in 29m 11s","accounts_in_message":[],"_revision_number":1},{"id":"3168300569330b35008f30e8a14319368222677b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-16 08:40:47.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"1dac100e07052b7161509d8ed5691601ed6e97c9","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-16 10:12:23.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"adf4caea7ab6408a64fe8604c806e4b34c52cd68","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 10:12:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2218f420577e2f15180879ff3a9500d07cc2a869","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:50:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"5e95e93e75c984978d4435c1a36e4a999dc14523","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:50:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b038eae78764a68a9591b2ac975ea08 : SUCCESS in 8m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53ed141a8dce4249aef077fd353800e5 : SUCCESS in 8m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/456c9e29a3ed47c88e013f4ca913ca8c : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d77bb3ff7c6049b6af42080a44c29f9f : SUCCESS in 9m 59s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/58fc823d4cc74f2f8a25b6e823e0a84c : SUCCESS in 7m 19s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/c1697359dc7b479d94bf0a362b151fae : SUCCESS in 11m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/b2a8ef31e551408aa8a291e6cca19309 : SUCCESS in 35m 10s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/17ef90f7a10f4dde8e139ca9a2179b29 : SUCCESS in 1h 05m 45s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/01502020ef22461abd460e04653f8c83 : SUCCESS in 19m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3340004340e34012864c5352e67f9fcb : SUCCESS in 2h 04m 41s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a5a6b759670a49278d5a352a72f4483f : SUCCESS in 20m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/49f38257ea834cc1a99355f31e2b8805 : SUCCESS in 32m 02s","accounts_in_message":[],"_revision_number":1},{"id":"5fadad574a6a97e217ea24fd342d5a6f532d3f9b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:54:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5b5262322f6458c9fd6251d0ea2833f : SUCCESS in 1m 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0cc3703797bcf10a4e4c317234010d4a29128642","revisions":{"0cc3703797bcf10a4e4c317234010d4a29128642":{"kind":"REWORK","_number":1,"created":"2020-06-12 12:51:34.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/33/735333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/33/735333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/735333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/735333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/33/735333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/33/735333/1"}}},"commit":{"parents":[{"commit":"87c91cc615578c17ece2e2a2aa0b3a4d8787ec83","subject":"Switch to newer openstackdocstheme and reno versions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/87c91cc615578c17ece2e2a2aa0b3a4d8787ec83"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-06-12 12:49:49.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-06-12 12:51:28.000000000","tz":60},"subject":"CI: set previous_release to ussuri","message":"CI: set previous_release to ussuri\n\nThis controls which release we are upgrading from.\n\nChange-Id: I16c86ce90d2b5a689e69681f12d0dbcb5c4cd6be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cc3703797bcf10a4e4c317234010d4a29128642"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cc3703797bcf10a4e4c317234010d4a29128642"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
