)]}'
{"id":"openstack%2Fkayobe~735334","triplet_id":"openstack%2Fkayobe~master~I300dfdbc0f51d89c585bcefce4e4be40c2b85bda","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I300dfdbc0f51d89c585bcefce4e4be40c2b85bda","subject":"Remove python 2 cleanup","status":"MERGED","created":"2020-06-12 12:54:48.000000000","updated":"2020-06-16 12:56:13.000000000","submitted":"2020-06-16 12:50:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735334-1592311808190-8bc90565","meta_rev_id":"cf2d82f59bc1a8b6ca23002b0157bb8123971ff1","_number":735334,"virtual_id_number":735334,"owner":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-16 12:50:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-12 15:53:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-12 14:56:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-12 15:53:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-12 14:56:55.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-12 15:53:52.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-06-16 12:50:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"901c77648876b88389090d44dc13f7a1fa1d7312","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-12 12:54:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"597cf5e20ad445b2804bac36bcd1c1d5ce8b8188","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 14:30:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e29d809d31f4473800922daeaa39081 : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8adfa19ddf841d69dbf1c65c7fc257e : SUCCESS in 3m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e3a0313b66e40929c2a1773402ad422 : SUCCESS in 4m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c505d11d453454282b4ea86d7a7ff1a : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d17a900e63d43d78aa896df1f79c4be : SUCCESS in 8m 46s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/34a18e385f7040e797719ec95616ad76 : SUCCESS in 4m 14s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/33a1eaa8481346bfafbb5d880b402e28 : SUCCESS in 7m 35s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ca442a8a9873458b9cbd99eb63a2c129 : SUCCESS in 32m 11s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/81f5768f90ca40dc946aa90342f34120 : SUCCESS in 1h 13m 32s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/421fdfbf3d934a0795cbd8c70e44a9c4 : SUCCESS in 18m 48s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6c5b5562fc7a4f4d910c678d585df0d5 : SUCCESS in 1h 35m 10s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/792bb65ff5ab4f879dcc73c25705ed3b : SUCCESS in 16m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3b61351ac0f8478ebf8cd18b803c7e84 : SUCCESS in 29m 31s","accounts_in_message":[],"_revision_number":1},{"id":"45062e0dfdd6530c5185bfa7eae7612c7a521d36","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-12 14:56:55.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"40b4b79d80dc18656d4fe7440b855f0d53cebce8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-12 15:53:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4a145582c9bc14471e94119e9c88c5a7c274b20d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 10:12:48.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e1edab32b1fa2919bb1fea6df58541b20201181","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:50:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d03841a407fc4038b1cb2ead12499213 : SUCCESS in 9m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d37ffb8893c84cdf93c0051643b35ad3 : SUCCESS in 7m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c1dc0fb4fcb47fa98efd21fbdfc23a3 : SUCCESS in 8m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c013a6fcec2a49a194acc1ca56088bc3 : SUCCESS in 12m 01s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cb91117882794a73beba00ec33326a15 : SUCCESS in 7m 24s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/41639b05ee3d43a5a773efcd41b96dad : SUCCESS in 13m 44s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eb31b1e923084dfc83cc3b211083087a : SUCCESS in 43m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6fe15b2d55ac49c6a46eeb9d06756c44 : SUCCESS in 1h 13m 39s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/16056c505fec4db0bb437e77f4b0d816 : SUCCESS in 18m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/3c815e2409104992b4f0ba5b5db25329 : SUCCESS in 2h 02m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6b6577c6763b4e96828e54cfeba08392 : SUCCESS in 20m 50s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/21f317d664564a7f98aebb999f569499 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":1},{"id":"5957000d7dbb72c35186a2feb203111d0afdf781","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:50:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cf2d82f59bc1a8b6ca23002b0157bb8123971ff1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 12:56:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e749145935d24e2cb3ee0ee5eb9569dd : SUCCESS in 1m 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4caf460b5f3b949ea240b31c7ed8c30526e7836b","revisions":{"4caf460b5f3b949ea240b31c7ed8c30526e7836b":{"kind":"REWORK","_number":1,"created":"2020-06-12 12:54:48.000000000","uploader":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"ref":"refs/changes/34/735334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/34/735334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/735334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/735334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/34/735334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/34/735334/1"}}},"commit":{"parents":[{"commit":"0cc3703797bcf10a4e4c317234010d4a29128642","subject":"CI: set previous_release to ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0cc3703797bcf10a4e4c317234010d4a29128642"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-06-12 12:53:49.000000000","tz":60},"committer":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-06-12 12:53:49.000000000","tz":60},"subject":"Remove python 2 cleanup","message":"Remove python 2 cleanup\n\nThis is no longer necessary since the previous release, ussuri, supports\npython 3 only.\n\nChange-Id: I300dfdbc0f51d89c585bcefce4e4be40c2b85bda\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4caf460b5f3b949ea240b31c7ed8c30526e7836b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4caf460b5f3b949ea240b31c7ed8c30526e7836b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
