)]}'
{"id":"openstack%2Fkayobe~735365","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I0e3be19eed371853d80c401d6a46770a7d6dcbc5","project":"openstack/kayobe","branch":"stable/ussuri","topic":"feature-flags-stable/ussuri","hashtags":[],"change_id":"I0e3be19eed371853d80c401d6a46770a7d6dcbc5","subject":"Update feature flags","status":"MERGED","created":"2020-06-12 15:50:47.000000000","updated":"2020-06-15 12:02:04.000000000","submitted":"2020-06-15 12:00:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735365-1592222420966-2c7c96e1","meta_rev_id":"f51966252fbb7748c21a93917c3dd6f86f6fa1aa","_number":735365,"virtual_id_number":735365,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-15 12:00:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-06-12 15:50:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-15 09:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-15 09:55:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"date":"2020-06-12 15:50:47.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-06-12 15:50:57.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-12 15:50:47.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-15 09:55:40.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-15 12:00:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d4d131b1a4d88bbfd110bb715170458257aaa58","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-12 15:50:47.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1d2c82a42938fee2349950239487aba40e9eee52","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-12 15:50:57.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"ffc80f0c071f39f3a43eb5af9b9aa4fe89a78050","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 18:02:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff4374c4dbb14197a24be2f47d654829 : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adf357760c134450b9ad8bc902121206 : SUCCESS in 4m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0593a27bb08e4f8287cafc71b3b89fc9 : SUCCESS in 4m 03s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1a73dd0f75224b638700359a7fa7eaa1 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/576b0dff03fa4af3aa3068a262080b15 : SUCCESS in 6m 03s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/84b2eaa2106e42c5bd8688d6ae645195 : SUCCESS in 4m 27s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0b37ff8bfc424e0fbb354aa327b51804 : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d370d717c9da4ad2ae7e055238ada00b : SUCCESS in 34m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/65003c85cf8448448c670d8bffccd31d : FAILURE in 9m 41s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/41924d8a4ba04b4bad88b44dc8d82666 : SUCCESS in 15m 01s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/68f0ab12f0274eb79d8d3e3894615f6b : SUCCESS in 2h 10m 40s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5acbbf577c8f4301b0c0ef235023fd90 : SUCCESS in 20m 01s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cea5678c621e40bbb6d9e2732b2afe06 : SUCCESS in 26m 39s","accounts_in_message":[],"_revision_number":1},{"id":"f0f07499451c5cf76cdbfc3ddb3bfdf3b8a1616b","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-12 19:29:09.000000000","message":"Patch Set 1:\n\nrecheck\n\nrepo issue","accounts_in_message":[],"_revision_number":1},{"id":"79742ea17fc8ecc00b8c194dd7ef97f9a7a481d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-12 21:23:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe27f4e857524d18acc4f51ca28776ac : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5391249b22ab499195b7e558ea9fceaf : SUCCESS in 7m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d529769f9d414b329b7fbf6e506fdfce : SUCCESS in 8m 10s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0520fc8258d541b4a75d8b82ee2228ab : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cca02d96986404395b9fdd56a730b5a : SUCCESS in 6m 08s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a931f184ebb14bf8a46d881f5ed41822 : SUCCESS in 3m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8f4bfb715bfd44dabde911f07377947b : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8fcd065dece34377ba2461acadce65dd : SUCCESS in 32m 51s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/108ec39475ec40308531b752b9003cce : SUCCESS in 1h 00m 18s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/158ecfd4c3a9420581aa103fa8ab9057 : SUCCESS in 12m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/79d795ed18cb4407b9f002b43ac5f6f1 : SUCCESS in 1h 50m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/918011f70d4b4119b28befa33d771405 : SUCCESS in 17m 05s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/19f27877e20347879ec70e0581ed008d : SUCCESS in 25m 28s","accounts_in_message":[],"_revision_number":1},{"id":"de9eed960ab9b86efd85bcdf1cc30937c3f3dcf3","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-15 09:55:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a91fcff02f11c03cd57a67514b9fd47e28c9bb8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 09:59:23.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ee6b28526c46c0a1ca4c0efbef97541f5fbe2935","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 12:00:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b99b87be96764413b293c53cf6eb2d07 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5a0ef740aeab417eb703c2e1d4dd9d43 : SUCCESS in 5m 25s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0bd4b56812fa4e748fa0ae474f8efe4f : SUCCESS in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523c08bcd66c4565aaa306e4275bc74a : SUCCESS in 5m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/32fe60d78b4f43639162c7fb71fcd269 : SUCCESS in 6m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fa00d1d46b354c458c389f0e864b6943 : SUCCESS in 10m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f78640e5751149d3a77563c523a11f01 : SUCCESS in 28m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/f483563941634c6e81a7515f9ee1b712 : SUCCESS in 1h 06m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/5fc4820ddbe1498da1f8e087904e8498 : SUCCESS in 17m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/16f6e0e7d44c4b18a46a76391f0536d3 : SUCCESS in 1h 56m 03s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/99f5299705924a18a1c13f1832e76f8a : SUCCESS in 17m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e7e80dd673ee4f84a19d0313e429c68b : SUCCESS in 27m 02s","accounts_in_message":[],"_revision_number":1},{"id":"aacae038203ec05a435821aa0e4f4408913640e3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 12:00:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"ff7c28d56c41426725bea23a4a4b856d1d45e7bf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 12:02:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdccd6374951437eb0a31bceabbca25c : SUCCESS in 1m 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"892ce612602602f2b0a14c3eb3e3b1c860647182","revisions":{"892ce612602602f2b0a14c3eb3e3b1c860647182":{"kind":"REWORK","_number":1,"created":"2020-06-12 15:50:47.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/65/735365/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/65/735365/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/735365/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/735365/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/65/735365/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/65/735365/1"}}},"commit":{"parents":[{"commit":"a3c3e0b90451118509709a64312fb29637144786","subject":"Disable broken linting of literalinclude sphinx directive","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/a3c3e0b90451118509709a64312fb29637144786"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-06-08 11:40:45.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-12 15:50:47.000000000","tz":0},"subject":"Update feature flags","message":"Update feature flags\n\nThe flags in etc/kayobe/kolla.yml were out of sync with\nansible/roles/kolla-ansible/vars/main.yml.\n\nThis commit sorts them all alphabetically and adds a script to simplify\ngenerating the list.\n\nTrivialFix\n\nChange-Id: I0e3be19eed371853d80c401d6a46770a7d6dcbc5\n(cherry picked from commit 1e963a60ada175482c415ca2f6c46a5899a6939a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/892ce612602602f2b0a14c3eb3e3b1c860647182"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/892ce612602602f2b0a14c3eb3e3b1c860647182"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
