)]}'
{"id":"openstack%2Fkayobe~735566","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Ib6ab41a3ba320a1fe15d0d23561fad2fab7861e6","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"Ib6ab41a3ba320a1fe15d0d23561fad2fab7861e6","subject":"Fix seed VM interface ordering","status":"MERGED","created":"2020-06-15 11:51:35.000000000","updated":"2020-06-18 18:13:23.000000000","submitted":"2020-06-18 18:08:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"735566-1592503712368-324e7388","meta_rev_id":"ce650c0f5a7b9b8443674235f1aae4ef02dbb705","_number":735566,"virtual_id_number":735566,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 18:08:32.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-15 14:09:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-15 20:40:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-15 20:40:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-15 20:40:33.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-18 18:08:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"93d845a3b3476d1f2ef7ff88681c7fc0a4615be9","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-15 11:51:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3498adecd4961433919ef4120df477baa8ad495","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 13:49:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b65fd5c36033463d94f1d50ea2b996f5 : SUCCESS in 5m 12s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/29f2d33c2d2849a5a6ca20a5258e7d37 : SUCCESS in 5m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/78197bf62baa4b2d836e5f7f7b8d3890 : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f64f3561aab34910a7d327aa8ecc0c1d : SUCCESS in 5m 25s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bf23fb22e66445459c201f56665ac3c1 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28e9e63e4fbf4401be9c97b52e9cdfa6 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37ec087649e14058bd38182af1c8097c : SUCCESS in 3m 07s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae8e57a0c3584c499d3a2a2b9ccbaaf2 : SUCCESS in 7m 18s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/fddd56a71e6e4c68840cc3d7147813e8 : SUCCESS in 10m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1e400440aa0549e7a3f0a448416e06e0 : SUCCESS in 33m 51s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/d6e6cc0c6176475d8639e0cd67534a29 : SUCCESS in 1h 13m 12s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/478be280db4645569ead3cb7ab081168 : SUCCESS in 1h 56m 21s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/c4053fa960574fbcb2f4d3b9b361584e : SUCCESS in 24m 29s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/e23aee10b77740788eaf44ac0e5cdb92 : SUCCESS in 38m 50s","accounts_in_message":[],"_revision_number":1},{"id":"9150003e825d7552321bac585725ef0a86736c89","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-15 14:09:26.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fc8517ff8eb3530fdfcc1f62a26b9a069e8abf98","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-15 20:40:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ca2ea97ddf9e144e4991dfd0d587b375153eec05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 20:40:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1ec1378b42b23d445812cac84389fdc5507d9a22","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-15 22:38:08.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/e6fd3ca319274671833e66cdd31422c1 : SUCCESS in 9m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f1366ee1e4244e68cf5377a9919fd8e : SUCCESS in 6m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0a5f633b648453cb66c47d50b3667cd : SUCCESS in 7m 14s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/32b1936f9d3f49f9a894e200e2294bdf : SUCCESS in 8m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8ca3d47f2734396b2c4eccf178d1847 : SUCCESS in 8m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6d9967e15ce14d69bde60e4386ecade7 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/cbacfe9c6d684182967dba4ab66974f4 : SUCCESS in 7m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2f1fc1730f4942d7bae2effb85306c4c : SUCCESS in 11m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/73f3c167e7524380b2dd178f48e40e53 : SUCCESS in 38m 40s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/399e23ed8595410099a8a697b46a3e4f : FAILURE in 59m 43s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/848969ecb3ae4cd6a655deae18a9f9be : FAILURE in 1h 00m 04s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/78f36e1737fe45a183907ca47d1597fe : FAILURE in 23m 10s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5006b61d2afe4f438509035c3ad67985 : SUCCESS in 35m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8861a21f6e85640f53e9688578a63a2db335a1ff","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-16 07:49:25.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dccc5383ae371c4a453dafa398ddbb83977aed96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-16 09:34:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88c1461a20e9456a81b7526466f13cef : SUCCESS in 7m 06s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/656081c3703647d292cac93dac2ef5a2 : SUCCESS in 6m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dc000725b9cc453890709929c1c0fd81 : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53e09d2028dd4aca98a7522b0eb3d6d5 : SUCCESS in 7m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/de002dfa4c7f4b61bc4160c8b94bda56 : SUCCESS in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aef4749995640a2a1420cd8709efdff : SUCCESS in 7m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ac326c4a37341bb87630f33309c2f1f : SUCCESS in 5m 19s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/0181981b98024688bd587480578e912d : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0915c31d7d444b05917beee9fe75c69c : SUCCESS in 11m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/348c133e050142e58ec235d2010f2e9e : SUCCESS in 39m 40s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/c1486caae3f24a5380d1e58eda65a2d5 : FAILURE in 1h 07m 37s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/c021825d003a40439264be59413d24b2 : FAILURE in 1h 30m 15s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/dbab66fcd6124e67b01baadb3e633bd1 : SUCCESS in 22m 47s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/c39ee25ea80942d0bd6c6215c2172383 : SUCCESS in 45m 56s","accounts_in_message":[],"_revision_number":1},{"id":"c7db480c8249b88e6789b92e7f585f91d6d56dae","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 14:11:45.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"5c5ddb94fce8f9d297163d774922a0f6cb857043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 16:17:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8aa9ef6b1dd14b1ea4994416f46108f0 : SUCCESS in 5m 44s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/0961a994898844b08770b04a142533bf : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/214efff8c16f4d0899add0eaaa74d98f : SUCCESS in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75a884a4a0634dc1bd417724a39f8714 : SUCCESS in 5m 02s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/21e77a3bbfc649378b1e419e927692e1 : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29c6d25b6924413aa9eaaf095829fe6f : SUCCESS in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec156f02c9b847d4b13232684880425f : SUCCESS in 4m 32s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7ac86b08d9ed4019a7c30c690b3f6b0d : SUCCESS in 5m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ff6030fd039d4eb0842718becacbb035 : SUCCESS in 8m 40s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ab82d82b75f843648a6a94060cb68b54 : SUCCESS in 34m 35s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/05aeb694edc341ad9d68697cb6f2ad21 : SUCCESS in 1h 05m 05s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/ce51e32b391a456bb190ed9a4a2e73cd : SUCCESS in 1h 56m 19s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0859fefb69e64ffb860a47cd538a1262 : SUCCESS in 27m 40s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/5634373983154f0ab27241a0a8b64a1a : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":1},{"id":"a88edb6384f5b1aebed946178289e37adde0f7b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 16:17:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0464b2ddff11e674c16ddffc2f4473f93c82ec61","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 18:08:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"18f09f1fc455840859e6a8355dbac23da0530ec3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 18:08:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/207d45ee52dc43d4b796af9d090ce45f : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/90b7b86a4e5d458ea2b6a45647b50a7b : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ff48efdac4e450daca351cb5333ebad : SUCCESS in 7m 26s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/c8d507d2c8ea42de93521af2c0c558cb : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44a629f1041f430786c42ec9b950ef98 : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/26a8a3d743444377b759a942737b07c0 : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb9973e53ac6460d89d0343cdbd7e5c9 : SUCCESS in 7m 05s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f1f4151fb1624c778e73f00ccd5aa515 : SUCCESS in 10m 54s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/01f8b47c3d1b4801ab230c3f281559c6 : SUCCESS in 26m 27s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5030ddf98a2b4f2fa2996da2b0dffbb0 : SUCCESS in 1h 01m 59s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/7f58c5d7343f49e08503ee99a36ef5a1 : SUCCESS in 1h 48m 36s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/8a6319d923db43d591c2675f73bf33b2 : SUCCESS in 26m 27s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d135d8cab4d84167b914ab0c5641b512 : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":1},{"id":"ce650c0f5a7b9b8443674235f1aae4ef02dbb705","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 18:13:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a454f1fe25dc405ab302ffaf88ccbb5e : SUCCESS in 4m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7761a1e6caac44339f15d995af430d1b : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"098612dd7bbf189b40ad20eab8c85c78fe9a556a","revisions":{"098612dd7bbf189b40ad20eab8c85c78fe9a556a":{"kind":"REWORK","_number":1,"created":"2020-06-15 11:51:35.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/66/735566/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/66/735566/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/735566/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/735566/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/66/735566/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/66/735566/1"}}},"commit":{"parents":[{"commit":"7ea99f82dd686acca0ebd0cbfdee1eadadb1dcc2","subject":"Merge \"Fix concurrency issues while adding SSH keys to known_hosts\" into stable/stein","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7ea99f82dd686acca0ebd0cbfdee1eadadb1dcc2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-08 08:53:12.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-15 11:51:26.000000000","tz":60},"subject":"Fix seed VM interface ordering","message":"Fix seed VM interface ordering\n\nCurrently the ordering of network interfaces in the seed VM is\nnon-deterministic. This happens because we apply the \u0027unique\u0027 filter to\nthe network_interfaces list, which does not guarantee a deterministic\nordering. This list is then transformed and passed to the\nstackhpc.libvirt-vm role.\n\nThere are two consequences of this:\n\n* it is not possible to determine which interface names should be used\n  prior to creating a seed VM\n* if a seed VM is recreated, the interface ordering may change\n\nThis change fixes the issue by sorting the network_interfaces list\nalphabetically before it is transformed and passed to the\nstackhpc.libvirt-vm.\n\nA new \u0027seed_vm_interfaces\u0027 variable is also added, which allows for\ncustomisation of the VM\u0027s interfaces - potentially allowing for more\ncomplex setups such as trunked VLANs.\n\nStory: 2007259\nTask: 38621\n\nThere is a second issue, which is that if the seed VM has a\nnetwork interface not configured with a gateway, cloud-init will fail to\nconfigure the network interfaces on the host. This has been observed on\nCentOS 8, but is probably more tied to the version of cloud-init, and\nmay affect CentOS 7. The following error is seen in the cloud-init logs:\n\n    KeyError: \u0027gateway\u0027\n\nThis change has been addressed in the jriguera.configdrive role, and\nthis change updates the version used in requirements.yml.\n\nStory: 2007769\nTask: 39993\n\nChange-Id: Ib6ab41a3ba320a1fe15d0d23561fad2fab7861e6\n(cherry picked from commit 7bcc5579e2de1735deaf970c5093a4ce629a37f0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/098612dd7bbf189b40ad20eab8c85c78fe9a556a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/098612dd7bbf189b40ad20eab8c85c78fe9a556a"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
