)]}'
{"id":"openstack%2Fkayobe~736101","triplet_id":"openstack%2Fkayobe~master~I3c921f5a558ae57ecc22fc6fc1959fb621c57709","project":"openstack/kayobe","branch":"master","hashtags":[],"change_id":"I3c921f5a558ae57ecc22fc6fc1959fb621c57709","subject":"Remove the neutron-fwaas since it retired","status":"MERGED","created":"2020-06-17 05:35:50.000000000","updated":"2020-07-01 23:21:51.000000000","submitted":"2020-07-01 23:20:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736101-1593645634520-72369fb8","meta_rev_id":"229ad92c9fd6949166cf82a536e7fcefc7c33af6","_number":736101,"virtual_id_number":736101,"owner":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-01 23:20:34.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"recommended":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":1,"date":"2020-07-01 08:24:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":2,"date":"2020-06-30 08:32:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-07-01 21:21:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-07-01 21:21:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},{"value":0,"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-30 08:32:27.000000000","updated_by":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"reviewer":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"state":"REVIEWER"},{"updated":"2020-07-01 08:24:05.000000000","updated_by":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"reviewer":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"state":"REVIEWER"},{"updated":"2020-07-01 21:21:02.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-01 23:20:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1edc21a2f506cd31abbb1323a8cf6cf0eb0a194c","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-06-17 05:35:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cda82a5fc88e697cd21f34981205e47d50b8d1cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 07:37:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57eb02d4604a4d49b57bfe1f4880ff6e : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af2621d9ddb24f70a34a43bebb21269d : SUCCESS in 5m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/651355d4fd97431e808226cf8d52ed19 : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c58e285d9f86496198708cebb79ca1cd : SUCCESS in 6m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1b557e695f43fda9a999ce0d7a5f46 : SUCCESS in 8m 24s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/22efcbbe01034e1f8be97c9aba700f94 : SUCCESS in 4m 15s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b6d39a6f82014588b5bff7c8dc14b540 : SUCCESS in 8m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/5e599b0d4e7e46fda843908a4247fd18 : SUCCESS in 33m 38s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/beb9784f0a394845938e147d60cbed27 : SUCCESS in 1h 26m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1608e2f83e2446d2b22dabcfa5eb832b : SUCCESS in 15m 22s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e605eb3f3ddb4c168148d00c65a77c14 : SUCCESS in 1h 56m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/505e1579fb994b2c80e61a82c6a89e39 : SUCCESS in 17m 27s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/344a8272b6a741049f193367308242e3 : SUCCESS in 28m 52s","accounts_in_message":[],"_revision_number":1},{"id":"8bc1d8e973efdf1c96c91ba9a5fb707d65094716","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-06-20 15:31:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"fe94e2e1dc72e113ba9afe37e50271ca4be2f5f4","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-06-20 15:32:54.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f8e2e1bfc062234095aa574d994a3cf9fd948f9e","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-06-20 15:33:35.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"cc90105a66d777f6db100e02765bb1e9f7500718","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-20 17:29:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b62325b99954e55bf889b838db5f780 : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c933f83d3964d0e8756e86ac646caab : SUCCESS in 3m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/594bbfb5db5d4fafbaebe4f8b0fb2ba1 : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/394f2ed8c0184ea79fff4d424024852c : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/134065ce402c441d8c9a4569556b8eb3 : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb6487957c194cf3acd524c69aa2e751 : SUCCESS in 4m 30s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/061f4c5916ac4775b186e47447298545 : SUCCESS in 8m 50s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d10d03f4305248cf84fdf078029405c0 : SUCCESS in 33m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/33be8bb7f07e4f93a8e2bd01745a43a5 : SUCCESS in 1h 07m 13s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/25a7410cfcad497f838d7f1c347a8423 : SUCCESS in 17m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a6a367af8b4240e5b255fdd99d2f135a : SUCCESS in 1h 55m 23s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7eaa9e8c1a8240b8aba8222484219535 : SUCCESS in 18m 20s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9833009b737a4a5fb874a8f29f86d783 : SUCCESS in 24m 56s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/b4d2bb9871ac41168a499eda75796a8d : SUCCESS in 10m 53s","accounts_in_message":[],"_revision_number":3},{"id":"e31c503fdc8f3384c75fcd23c14f1468408de66a","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-22 05:22:32.000000000","message":"Patch Set 3: Code-Review-1\n\nPlease add a release note.","accounts_in_message":[],"_revision_number":3},{"id":"82a71eb91df11764b527e6c29e138abd1eb2fe58","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-06-23 15:30:11.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"fc46c971b3a603f3a5db419f3894cca81120fbd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-23 16:57:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41a410061c6e49c7a1864e5dfb783f90 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eddf542002d94ee99243ab8e248fb9f5 : SUCCESS in 5m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a6500fb7b9f94619a438d6ceff70e4d8 : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/804f544f46f2437ebdad5c56fa7051c0 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14c585ea27e94cddbbf642bebf44ac67 : SUCCESS in 9m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce3ead2f2c254c49b984525533a47b25 : SUCCESS in 3m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/10db85eacc9f43f3a75e1bbedb29c7ef : SUCCESS in 4m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5c695ea9b6c948888e869dff541fab4f : SUCCESS in 9m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/65a9d10262114da2bcaf3389ceb0263f : SUCCESS in 33m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9089153acc5043c8b3a0a7e10893a107 : SUCCESS in 1h 09m 26s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a74264d8553d47568df8673c43db5d38 : SUCCESS in 16m 27s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/97174dbeb68d49848c4fb3461b338685 : FAILURE in 1h 25m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/e1cdb48155a4444aa324628c56b2c475 : SUCCESS in 18m 17s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e3f1e34fe3964a869bad3733cf51509e : SUCCESS in 30m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/8668a130b3ce4c67a3758af043191efe : SUCCESS in 11m 29s","accounts_in_message":[],"_revision_number":4},{"id":"083bd92083f03b80719e236d7a59b7b7f89bbe06","author":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"date":"2020-06-26 03:48:59.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"bb73bb903aca89227290dc7572a8caa93b31b04c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-26 05:08:44.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4360bbe01cfe4f47978c952357488bf8 : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44c613d94afd43a0a24cd12711d005d8 : SUCCESS in 3m 41s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b6caf1a89ee424ab0877efe4ce92be6 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7a44adb6c1094249901772518c724608 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91181231ef034e53a579086e25e046ca : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/103b7532983149d0b276367c8e9d9a07 : SUCCESS in 3m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ee4a255b2fdc4f45be6dbc4c96858637 : SUCCESS in 3m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8df3105bf311459d839338a90d78c3a1 : SUCCESS in 6m 18s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fe46c30452b2412e9e0d585a41624849 : SUCCESS in 24m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/d6b9f31f693343bca43239318ea83cac : SUCCESS in 47m 34s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/caf9f825757d4606b54aae1e2483bfdc : SUCCESS in 12m 06s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6410c1d0588040b99e217db656918230 : FAILURE in 1h 18m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/8460266255bc48e68c780aaf76294bd7 : SUCCESS in 13m 44s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ecc109e35f9d435e81eec4465257f71b : SUCCESS in 20m 04s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/47b0e06dc45f4a249f2f006f24ff337d : FAILURE in 7m 57s","accounts_in_message":[],"_revision_number":4},{"id":"830307820d7536c193af709b741d986ada44fb99","author":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"date":"2020-06-30 01:39:03.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"6b5c4bce40ba93012863c2568365c9711910b411","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 03:53:20.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3976e0aa129945d2bf757eeb84c004aa : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e108e99155ce4584a92e26d91dc65600 : SUCCESS in 4m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f8520b2a68440d1bd4a0309606b2eb4 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65f0c0a3cfec4bf4b108ca8691073a4c : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e11495257b14e91b0b616eeb03b5420 : SUCCESS in 8m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac9db5968e3b4b1cbd1f459fef4ef2f0 : SUCCESS in 3m 18s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/381d17e892fa4e92ad2af1516633edee : SUCCESS in 4m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/832e74e1fc504a8e9c95e2f53ad1331e : SUCCESS in 9m 06s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/95c1d0c587fd4d7e86a8f6f5ae087d3d : SUCCESS in 35m 14s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/0b0bf31d6e8745d1bb24c0bdf9b0cd0f : SUCCESS in 1h 19m 31s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e80652fd99d04eab9dce5049be2727b6 : SUCCESS in 16m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/813adfdd7e3b4de488ae6a944f270beb : FAILURE in 2h 05m 21s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/38186d5d47c744ae881b664f8b8a125b : SUCCESS in 24m 34s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cb7c6c9537414db8917f72532fe10aaa : SUCCESS in 27m 57s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/f9b186d4acff4b50a2a63d9894b9628f : SUCCESS in 10m 20s","accounts_in_message":[],"_revision_number":5},{"id":"eeb83f33227bd74486d0e8597a3317197c463b5a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-30 08:32:27.000000000","message":"Patch Set 5: Code-Review+2\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"56f9121fe9e63dba6dfd466eb27b1054c402bf3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 10:40:06.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d03910d01664496e96814da4469d17a5 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6a34fdb420c74b2da6900ff1c8ce8bfd : SUCCESS in 6m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/febf6eb3259f45549b201a776baa32fa : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19dce71931094de6ae064da868ae7c26 : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90cecb951438471b979a485ab6ba9e38 : SUCCESS in 8m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe3994e1171c4ea988cd75c59f6096ec : SUCCESS in 4m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/fc0c3ff17f2440d793dd508cb3473d5d : SUCCESS in 5m 22s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/3b5fed705d724793b5b73f5c6b1b06d6 : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/86df37b84a5a4a28943ec3ae1f1ed03f : SUCCESS in 43m 05s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/3871ba1ea5d9436caf84e37c05d0339e : SUCCESS in 1h 10m 40s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/bdf11e521080409f8a1bb3e026763a35 : SUCCESS in 16m 40s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b9900700c27d4288a1efeca241c161e9 : SUCCESS in 1h 58m 04s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/0e785162114d4bf2aa0fb7233faffacf : FAILURE in 18m 40s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d3da310068004f90a1bb7faca9c3e3e3 : SUCCESS in 22m 46s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/da4d94eba8b84b49a7e96b06334cabfd : SUCCESS in 11m 02s","accounts_in_message":[],"_revision_number":5},{"id":"757d8a205f49b709f1a152b75af12a111b3d762d","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-30 12:50:45.000000000","message":"Patch Set 5:\n\nrecheck\n\nFailed to download remote objects and refs:  fatal: unable to access \u0027https://opendev.org/openstack/ironic-python-agent/\u0027: OpenSSL SSL_connect: SSL_ERROR_SYSCALL in connection to opendev.org:443","accounts_in_message":[],"_revision_number":5},{"id":"4bf88ef512d15751a7242bdc6a2b6f8a7ae0a2b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 14:42:49.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc54942cb8424e9b8ee8e4d325015717 : SUCCESS in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37fb546364d7411bae6633c2e6719546 : SUCCESS in 5m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/484306b778714c85abaf584d7f4a9853 : SUCCESS in 9m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/190409257d8e4d919b130002febec706 : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e79780ca8fa4e6a922ee81c01dfc9a8 : SUCCESS in 13m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3bae0e98c4c4cf29843e9fbd4c94868 : SUCCESS in 4m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/661465b184d74b78b78560d3068907ec : SUCCESS in 6m 23s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/13bf1db1a7c94f14a62dcd106b0e19bd : SUCCESS in 10m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/228dc53f434646e697a22666e4641c47 : SUCCESS in 30m 37s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a52cb75fb8cd40d4918dfb425775d6e4 : SUCCESS in 1h 10m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/6a623f8be1b7431e8184a611f905803a : SUCCESS in 16m 52s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/58a5de80f76e437181b873b9f7add1ba : SUCCESS in 1h 50m 39s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7d0ca8422b0a416c99b13dd231c40105 : SUCCESS in 21m 36s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1d10dfa3e08d4ef6b0dabb79751e564f : FAILURE in 19m 27s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/e49a214bb4844c7a93a1698526ddef3e : SUCCESS in 12m 15s","accounts_in_message":[],"_revision_number":5},{"id":"c90baa57be28f40aa9543ee473c6bd1e3602bc43","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-30 15:09:24.000000000","message":"Patch Set 5:\n\nrecheck\n\nFailed to download remote objects and refs:  fatal: unable to access \\\u0027https://opendev.org/openstack/ironic/\\\u0027: Failed to connect to opendev.org port 443: Connection timed out","accounts_in_message":[],"_revision_number":5},{"id":"da5d96e0b65725a9d34de01478eedb1782ce804e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-30 17:18:56.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8a7525fb85004cd28ed31112fc297b3e : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979ca2b70c4c47a9b468d462f32573c8 : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/53321e78926d4fd089550a9f56beb35d : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/87b7dbd39b4d491f9f05675d392afe6d : SUCCESS in 8m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b945fe39caa04750afd79548d0e384e4 : SUCCESS in 10m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/186416970f7e4185b85dda07b078ffcd : SUCCESS in 5m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/3843b5af15d6440c866aff1622b79361 : SUCCESS in 6m 07s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/9398f286a0c34ff188f8c1231c2b45ca : SUCCESS in 8m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7460599178fe48f2a6ada3ba84310a4c : SUCCESS in 37m 55s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4829394047fb46d495eeb4acdd220633 : SUCCESS in 1h 12m 23s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/47eca6b0865141ec944cc1a50df17cae : SUCCESS in 19m 26s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1cb95a87d3dc4a5b81c007ce7f55ce8a : SUCCESS in 2h 01m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/f821253d47ff4ed39cff84cf6500c02f : SUCCESS in 19m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e28d5d02d12347dca6a82639deb56b59 : SUCCESS in 23m 48s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/760f16cf9ad54f669bc470a6adfa13ba : SUCCESS in 11m 44s","accounts_in_message":[],"_revision_number":5},{"id":"182c0c0e5ec1a6a2e9e1deda6025bfad17478a10","author":{"_account_id":22165,"name":"caoyuan","email":"cao.yingjunz@gmail.com","username":"caoyuan"},"date":"2020-07-01 08:24:05.000000000","message":"Patch Set 5: Code-Review+1","accounts_in_message":[],"_revision_number":5},{"id":"768457ec7732bd71d5be87baa5bf4ec7f1129ec7","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-07-01 21:21:02.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"5dfd9b38998a615b475502bb42444961d14edced","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 21:21:14.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"2548c36e36be23236d860490aa161afa681d907f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 23:20:34.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7048d807d73147069a0a79026708c826 : SUCCESS in 7m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9d50a4b9d75047c5a3c2c737590d5912 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e8018c0412e4455a96fedf72b075b515 : SUCCESS in 8m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad3f35a6c2e1482e9c64fbd27448d6f0 : SUCCESS in 10m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/999850cacea24ea995d62bbc342ebeb3 : SUCCESS in 6m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e659c3ff53094e8dba474f7456e6e504 : SUCCESS in 5m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5201c4581d7a469f943ef3bac3d65db7 : SUCCESS in 11m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/13f882cd545e4e338ac819a25608eed1 : SUCCESS in 32m 27s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/02fc8454512942b390513c1dffa497c4 : SUCCESS in 1h 07m 19s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4bd4324164594a7286981e08450a2c83 : SUCCESS in 12m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6933cf04bb954001805b406b2e8296ec : SUCCESS in 1h 53m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ac581cc8302a45e3ac5bf231f88b5b20 : SUCCESS in 19m 47s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/53ec035884754ce988068c7bac9fc53e : SUCCESS in 25m 18s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/812e1e9c110343bbbb83fe1f171b27a1 : SUCCESS in 14m 25s","accounts_in_message":[],"_revision_number":5},{"id":"513acacd8a56f8869b92c0e7d75c193dfe3b64e9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 23:20:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"229ad92c9fd6949166cf82a536e7fcefc7c33af6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 23:21:51.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/487f7a3a456c4ba58f8bf762f2e8c488 : SUCCESS in 1m 02s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/89b6b9bb5c704ac289dcf4946a833097 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"7fdacf475923b8a2906dd3e26f5aaede7cb88383","revisions":{"ee308963734eaa041fd7019f9ba4049e14394a74":{"kind":"REWORK","_number":1,"created":"2020-06-17 05:35:50.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/01/736101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/736101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/736101/1"}}},"commit":{"parents":[{"commit":"0734e55ec051e5b0544626c0e15296c293fc1d83","subject":"Merge \"Update feature flags\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0734e55ec051e5b0544626c0e15296c293fc1d83"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:34:49.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:35:25.000000000","tz":480},"subject":"Remove the neutron-fwaas since it retired","message":"Remove the neutron-fwaas since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I3c921f5a558ae57ecc22fc6fc1959fb621c57709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee308963734eaa041fd7019f9ba4049e14394a74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ee308963734eaa041fd7019f9ba4049e14394a74"}]},"branch":"refs/heads/master"},"53cc5a47e1f1acff140ad7429e743bcec17fe567":{"kind":"REWORK","_number":2,"created":"2020-06-20 15:31:18.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/01/736101/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/736101/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/736101/2"}}},"commit":{"parents":[{"commit":"791c219f4d10982dbeebae0c68047af7ecb55b14","subject":"Remove remaining mentions of neutron-lbaas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/791c219f4d10982dbeebae0c68047af7ecb55b14"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:34:49.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-20 15:30:45.000000000","tz":480},"subject":"Remove the neutron-fwaas since it retired","message":"Remove the neutron-fwaas since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I3c921f5a558ae57ecc22fc6fc1959fb621c57709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53cc5a47e1f1acff140ad7429e743bcec17fe567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/53cc5a47e1f1acff140ad7429e743bcec17fe567"}]},"branch":"refs/heads/master"},"0425b68a48342b084999f58b3a56d4e6453294cc":{"kind":"REWORK","_number":3,"created":"2020-06-20 15:33:35.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/01/736101/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/736101/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/736101/3"}}},"commit":{"parents":[{"commit":"791c219f4d10982dbeebae0c68047af7ecb55b14","subject":"Remove remaining mentions of neutron-lbaas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/791c219f4d10982dbeebae0c68047af7ecb55b14"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:34:49.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-20 15:33:08.000000000","tz":480},"subject":"Remove the neutron-fwaas since it retired","message":"Remove the neutron-fwaas since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I3c921f5a558ae57ecc22fc6fc1959fb621c57709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0425b68a48342b084999f58b3a56d4e6453294cc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0425b68a48342b084999f58b3a56d4e6453294cc"}]},"branch":"refs/heads/master"},"299ddd0353aa0e7406bf3041d41f727a8b02e991":{"kind":"REWORK","_number":4,"created":"2020-06-23 15:30:11.000000000","uploader":{"_account_id":30356,"name":"gugug","email":"gu.jin@99cloud.net","username":"gujin"},"ref":"refs/changes/01/736101/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/736101/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/736101/4"}}},"commit":{"parents":[{"commit":"791c219f4d10982dbeebae0c68047af7ecb55b14","subject":"Remove remaining mentions of neutron-lbaas","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/791c219f4d10982dbeebae0c68047af7ecb55b14"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:34:49.000000000","tz":480},"committer":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-23 15:29:43.000000000","tz":480},"subject":"Remove the neutron-fwaas since it retired","message":"Remove the neutron-fwaas since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I3c921f5a558ae57ecc22fc6fc1959fb621c57709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/299ddd0353aa0e7406bf3041d41f727a8b02e991"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/299ddd0353aa0e7406bf3041d41f727a8b02e991"}]},"branch":"refs/heads/master"},"7fdacf475923b8a2906dd3e26f5aaede7cb88383":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2020-06-30 01:39:03.000000000","uploader":{"_account_id":28614,"name":"jacky06","email":"zhang.min@99cloud.net","username":"jacky06"},"ref":"refs/changes/01/736101/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/01/736101/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/01/736101/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/01/736101/5"}}},"commit":{"parents":[{"commit":"931fa4830d681281ff189bed15fbc9e0554428c0","subject":"Fix condition of placement image build","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/931fa4830d681281ff189bed15fbc9e0554428c0"}]}],"author":{"name":"gugug","email":"gu.jin@99cloud.net","date":"2020-06-17 05:34:49.000000000","tz":480},"committer":{"name":"jacky06","email":"zhang.min@99cloud.net","date":"2020-06-30 01:39:03.000000000","tz":0},"subject":"Remove the neutron-fwaas since it retired","message":"Remove the neutron-fwaas since it retired\n\nDepends-On: I561504160e5548c54d1af31821c3366ab34cf0ec\nChange-Id: I3c921f5a558ae57ecc22fc6fc1959fb621c57709\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fdacf475923b8a2906dd3e26f5aaede7cb88383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/7fdacf475923b8a2906dd3e26f5aaede7cb88383"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
