)]}'
{"id":"openstack%2Fkayobe~736241","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I2268a1ddf9a2870c713f32a40689e1686365aabd","project":"openstack/kayobe","branch":"stable/ussuri","topic":"story/2004759-stable/ussuri","hashtags":[],"change_id":"I2268a1ddf9a2870c713f32a40689e1686365aabd","subject":"Add seed VM provisioning CI job","status":"MERGED","created":"2020-06-17 14:37:02.000000000","updated":"2020-07-10 18:30:08.000000000","submitted":"2020-07-10 18:27:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736241-1594405621772-de16a12a","meta_rev_id":"61bd4bb1d163c14940b285bbbf3f702e8e557c2f","_number":736241,"virtual_id_number":736241,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-10 18:27:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-19 14:15:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-19 12:29:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-07-10 16:41:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 12:29:18.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-07-10 18:27:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d64a3ad6297b720f95860a3bb0469ea75967300","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 14:37:02.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7d50e21ddb464a6b4218d6e463ce4c60a2fec3a0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 14:37:44.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"222c980dd6748cf97918373e1c1cc06b9de1e1f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 17:07:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bcbbb598e21342aa8bf326d20cc4ee3d : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4bdadb993444a2eabaca5265ba937cb : SUCCESS in 7m 18s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/36bd33ad776948f2a0ea3979ff5b46b2 : SUCCESS in 7m 50s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1fd89291fd484798af95b30f9f0e21a9 : SUCCESS in 7m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb059a4685254967ba70bbe73adea690 : SUCCESS in 10m 54s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/339fe8cbc959415985d6185970bc980a : SUCCESS in 7m 11s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/5ba165dd65f74df99230738d2832b1b4 : SUCCESS in 11m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3f90a5dffc614cfa81654df34199b70a : SUCCESS in 28m 42s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5e1b5a0b672e48f682ffef7128adaa56 : SUCCESS in 1h 17m 52s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/97064d4f994448e7becfc1fdae9c387d : SUCCESS in 14m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/2d92db7de2ab48729706c896850dcfdb : SUCCESS in 2h 23m 37s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/81e12088c48d41dabd576dded72830a6 : SUCCESS in 23m 36s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b668da77a54e498dabab5e5249331263 : SUCCESS in 29m 37s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/d3b5e6bad28344478568861b0fe6b5e3 : SUCCESS in 14m 47s","accounts_in_message":[],"_revision_number":2},{"id":"8bdb4871eb14f59d719dcbaad9fe1776a6635275","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 12:29:18.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"e5b8b3b8cc930b8f6ca4ba93e7542b992de69a7d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-19 14:15:50.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ed789fece7854665713a18639b64e192044ff2f3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-10 16:40:58.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"40cbafe429cb8b6e0a54de979af7a4ec5ba6d195","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-07-10 16:41:03.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ca6bf471f827d78a328ff3b3348eafd4202191a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 16:42:27.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5f1d01ef6d622c2ec9fa0fd5cf16b34243fa4487","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 18:27:01.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1530acb87784831baa282d7bfded4c4 : SUCCESS in 6m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/836104eabdd147fea0ab0895ef74abd8 : SUCCESS in 5m 24s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d44341178f064d86ae29460fd7ea023b : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2846c9832ae74e24a548a93e6b2e0b72 : SUCCESS in 8m 44s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/7da8a1d8be624c2a914062cab011fd52 : SUCCESS in 10m 41s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/23072c9ecf4544aaa393278cd5117717 : SUCCESS in 10m 37s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d76f17f48e8d41febcb1fc24e4c57faa : SUCCESS in 39m 25s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/9e6e6c152d50400a8492052c77817920 : SUCCESS in 1h 15m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1ce884019ad0487fb5bfab4c9196ea30 : SUCCESS in 17m 19s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6562086d325f47f281d2e7ad104e7ed6 : SUCCESS in 1h 42m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ce13e13eb07b464ca81116f8c8abe3ef : SUCCESS in 17m 36s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/311cd61aa7b4488d9c900d1b78732fbf : SUCCESS in 29m 51s\n- kayobe-seed-vm-centos8 https://zuul.opendev.org/t/openstack/build/21cb9dddf3394012bd628750b9de364a : SUCCESS in 12m 51s","accounts_in_message":[],"_revision_number":2},{"id":"c8181708b7febc394806c42e627444687c8bc89e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 18:27:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"dd1c8514f4973ce585bfa0afcb4804d38a743c79","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-10 18:30:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9380ef305bb4672a6bf507ba7b21ccb : SUCCESS in 1m 15s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ec71150a90c145268e0eed489403831d6c85d099","revisions":{"0d064bb77af828d4a5eeb1052b093bc9f9b9f2a5":{"kind":"REWORK","_number":1,"created":"2020-06-17 14:37:02.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/41/736241/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/736241/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/736241/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/736241/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/736241/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/736241/1"}}},"commit":{"parents":[{"commit":"1b994378bbc9184dea0f1afd98cb4fe8a3bfb12b","subject":"Merge \"Fix seed VM interface ordering\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b994378bbc9184dea0f1afd98cb4fe8a3bfb12b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-17 14:37:02.000000000","tz":0},"subject":"Add seed VM provisioning CI job","message":"Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nThe job uses a Cirros image for the seed VM rather than the usual CentOS\ncloud image. This is to reduce bandwidth required to download the image.\nIt does mean that the resulting seed VM cannot be used as a seed, but\nnested virt would make this slow and unreliable anyway. Cirros does not\nload cdrom drivers by default, so we add the configdrive as a disk\nrather than a cdrom device.\n\nDepends-On: https://review.opendev.org/617161\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n(cherry picked from commit c16597aa2daf76d7dd31991c5db938eee291a098)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d064bb77af828d4a5eeb1052b093bc9f9b9f2a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/0d064bb77af828d4a5eeb1052b093bc9f9b9f2a5"}]},"branch":"refs/heads/stable/ussuri"},"ec71150a90c145268e0eed489403831d6c85d099":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2020-06-17 14:37:44.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/41/736241/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/41/736241/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/736241/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/736241/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/41/736241/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/41/736241/2"}}},"commit":{"parents":[{"commit":"1b994378bbc9184dea0f1afd98cb4fe8a3bfb12b","subject":"Merge \"Fix seed VM interface ordering\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1b994378bbc9184dea0f1afd98cb4fe8a3bfb12b"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2018-04-06 18:47:32.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-17 14:37:38.000000000","tz":0},"subject":"Add seed VM provisioning CI job","message":"Add seed VM provisioning CI job\n\nAdds the kayobe-seed-vm-centos8 CI job to configure the Zuul VM as a\nseed hypervisor, and use nested virt to provision a seed VM.  This\nensures that the seed hypervisor code paths are tested.\n\nThe job uses a Cirros image for the seed VM rather than the usual CentOS\ncloud image. This is to reduce bandwidth required to download the image.\nIt does mean that the resulting seed VM cannot be used as a seed, but\nnested virt would make this slow and unreliable anyway. Cirros does not\nload cdrom drivers by default, so we add the configdrive as a disk\nrather than a cdrom device.\n\nDepends-On: https://review.opendev.org/736240\n\nChange-Id: I2268a1ddf9a2870c713f32a40689e1686365aabd\nStory: 2001655\nTask: 6683\n(cherry picked from commit c16597aa2daf76d7dd31991c5db938eee291a098)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec71150a90c145268e0eed489403831d6c85d099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/ec71150a90c145268e0eed489403831d6c85d099"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
