)]}'
{"id":"openstack%2Fkayobe~736278","triplet_id":"openstack%2Fkayobe~stable%2Frocky~Id482ae69fc84b106d9d11a9a1693d94ce5f34ba5","project":"openstack/kayobe","branch":"stable/rocky","hashtags":[],"change_id":"Id482ae69fc84b106d9d11a9a1693d94ce5f34ba5","subject":"CI: don\u0027t run virtualenv on an existing virtualenv","status":"MERGED","created":"2020-06-17 15:52:20.000000000","updated":"2020-06-18 12:49:07.000000000","submitted":"2020-06-18 12:49:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736278-1592484547148-9c47ad73","meta_rev_id":"7a384d0b167b7ddd4bb8ab9facab7241757df869","_number":736278,"virtual_id_number":736278,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 12:49:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-18 11:14:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-18 11:14:10.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-18 10:05:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 06:09:53.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-18 12:49:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"31851dfbdcca7121c1ebe5cb99143d665580e069","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 15:52:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a3a50f3dfe6fc419b8f1eb0b5cbef5cbb093f492","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 17:04:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/c8a2a04e27da457a99bbf3f07a4d4961 : SUCCESS in 7m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58b9d454609249edb44f8cee87019da2 : SUCCESS in 6m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b484874a45044339a4a9151828bd2ddc : SUCCESS in 7m 11s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/f1efbe3f6f6b4f968187e2ccc6d9be80 : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e522033bafd458e89993c637eca9fad : SUCCESS in 7m 05s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/182a8d7f5a7a49d28dea09d76596ef7e : SUCCESS in 7m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bdcf06debda741f79ad23a252b3a9e51 : SUCCESS in 7m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/f1459948552b4532a6fd92962bea2212 : FAILURE in 7m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/abfb0ceedf824f93ab17ea7a7928c466 : SUCCESS in 33m 26s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5b8dbb72ce9645d48c490933e5292592 : SUCCESS in 1h 07m 01s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/91fbac3f81df4cebabc9e1f5e697245d : FAILURE in 52m 58s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/18ead801713d488488a354e5493a3847 : SUCCESS in 18m 24s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/f4219129990d448c85955ea595182f64 : FAILURE in 16m 30s","accounts_in_message":[],"_revision_number":1},{"id":"e66dbde9254793d1b07551a2e62acc23b5f6ebb3","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 17:53:42.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"a44737f1b49dc1208212357b892df766b9c09201","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 06:09:53.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ab5b1d658e938aeb48ee60d4101544f3e5e92c05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 07:34:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/bca728215d034cc59dc57e86a71a106c : SUCCESS in 8m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2017d6698744ca4b2028c9b9b4cc445 : SUCCESS in 6m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a55d3709d7984a4284509d075e20642d : SUCCESS in 7m 26s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/475de92dbddc482684104300c10041d9 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3510ad412774021ae3b601fbf9b5b7f : SUCCESS in 6m 55s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e77411a0ec614d86bc7ae160991e53f4 : SUCCESS in 8m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/d8fa02fe6be04211b8807ca51b032398 : SUCCESS in 8m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b93c6c99d9124fe6a0bbb12e890be8d3 : FAILURE in 9m 00s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ac9a3a8e681445ccbce142edbf03becd : SUCCESS in 29m 14s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/ac3090a4afce4a8db17a65eabd7dc4f2 : SUCCESS in 1h 16m 48s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/fe8ae45743674d858c6b0235126cd38f : FAILURE in 1h 15m 41s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/bda63f81014649bd9b6b110456d507d6 : FAILURE in 13m 52s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/d4f907e4996d44f58acd4b3a9537a587 : FAILURE in 12m 55s","accounts_in_message":[],"_revision_number":2},{"id":"fdad2a65cfa1293f3c82f951e4e207c736626222","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 09:39:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"80f3eb0ad112563382953f23bfa3cdbe590552a6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 10:02:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"df38e1557a7734f8a3906a2fc981b2a097a75eac","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 10:05:36.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"224b838a53bd1e84c982dc218d18175032b58f0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 11:07:45.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/3cbadb0e4589409d81c7eba549d4b21c : SUCCESS in 8m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1971d30ec67c4c008689d752fc0e2095 : SUCCESS in 7m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c4081e397b84e68a143e56989e9d4c5 : SUCCESS in 8m 21s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/c8ee9dbea0624592a24bf2ed9cb60e51 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0c2438cfc3e43ee9ea18bf1c7b9c871 : SUCCESS in 4m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/9a1adbc2b0e24a5789f71d9564ec3e6d : SUCCESS in 6m 40s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/62824d7f7d6e4fc49ac64585ce32c9eb : SUCCESS in 7m 21s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2fc42e3156a94ea4a1a2ce2600c13d50 : FAILURE in 8m 39s (non-voting)\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6efe8dc415484b208d45ac0709cd936e : SUCCESS in 28m 46s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/5bd2376dbf9747b28743bc175089b53e : SUCCESS in 59m 54s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/616adc2bf54748e09a9c3c3d791d3017 : FAILURE in 47m 06s (non-voting)\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/46fec17a3c35491c9041dc779823335d : SUCCESS in 22m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/10c43d4c41324d51b43fa3e0812dda01 : FAILURE in 15m 15s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"8ca6721c12cee73be15d1347a4df222e3ff7342d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 11:14:10.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"f0678fd89df5844a60f2d4206e2280b3c419a3da","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 11:14:30.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"54de96b75c8df0c0cc7ffb6ae8f1d20dceed9efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:49:06.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs https://zuul.opendev.org/t/openstack/build/80cbcc1d4f7b4236a4156d31fa12bda7 : SUCCESS in 7m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e95f8856e0194dfd8b38e3e7c59c0978 : SUCCESS in 5m 02s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/92ba82bb6b3c4bd29dc1e4dd0c7f8efc : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd4a612e19ae4fbabbfc4649d7cce901 : SUCCESS in 6m 41s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bc59e18780a6467088d0c088ee24acd4 : SUCCESS in 7m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a2e547eaa9c642219b0e41596639181d : SUCCESS in 8m 53s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/987830b6492a4ad194ea94e1b4045f48 : SUCCESS in 28m 16s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/dba5ed94668944a8b3224a6958b3d91f : SUCCESS in 55m 34s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/0d3f9a664f21490aa8199194c77c641b : SUCCESS in 26m 43s","accounts_in_message":[],"_revision_number":5},{"id":"7a384d0b167b7ddd4bb8ab9facab7241757df869","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:49:07.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"6dd11cea73e0e2df4d65ecc3744e4abc68ff195f","revisions":{"bacef705de5673c259021b3edfad6e841d4545dd":{"kind":"REWORK","_number":1,"created":"2020-06-17 15:52:20.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/736278/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/736278/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/736278/1"}}},"commit":{"parents":[{"commit":"5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59","subject":"Use ensure-docker role instead of install-docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-16 13:53:06.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-17 15:51:33.000000000","tz":60},"subject":"CI: don\u0027t run virtualenv on an existing virtualenv","message":"CI: don\u0027t run virtualenv on an existing virtualenv\n\nThe seed upgrade job is currently failing on CentOS 7:\n\nTraceback (most recent call last):\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 2327, in \u003cmodule\u003e\n    main()\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 712, in main\n    symlink\u003doptions.symlink)\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 924, in create_environment\n    site_packages\u003dsite_packages, clear\u003dclear, symlink\u003dsymlink))\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 1369, in install_python\n    os.symlink(py_executable_base, full_pth)\nOSError: [Errno 17] File exists\n\nThis is likely due to changes in the nodepool infra image which removed\nthe default installation of virtualenv.\n\nThis change avoids running virtualenv on an existing virtualenv.\n\nIt also installs the python six package in CI, which is required by\nTenks.\n\nChange-Id: Id482ae69fc84b106d9d11a9a1693d94ce5f34ba5\n(cherry picked from commit 050042ceccff5d87ba67e22f36f6c6ce8df7a663)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bacef705de5673c259021b3edfad6e841d4545dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/bacef705de5673c259021b3edfad6e841d4545dd"}]},"branch":"refs/heads/stable/rocky"},"8d9cc2f22535134aece35c731ede7373f06682bf":{"kind":"REWORK","_number":2,"created":"2020-06-18 06:09:53.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/736278/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/736278/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/736278/2"}}},"commit":{"parents":[{"commit":"5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59","subject":"Use ensure-docker role instead of install-docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-16 13:53:06.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-18 06:09:45.000000000","tz":120},"subject":"CI: don\u0027t run virtualenv on an existing virtualenv","message":"CI: don\u0027t run virtualenv on an existing virtualenv\n\nThe seed upgrade job is currently failing on CentOS 7:\n\nTraceback (most recent call last):\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 2327, in \u003cmodule\u003e\n    main()\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 712, in main\n    symlink\u003doptions.symlink)\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 924, in create_environment\n    site_packages\u003dsite_packages, clear\u003dclear, symlink\u003dsymlink))\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 1369, in install_python\n    os.symlink(py_executable_base, full_pth)\nOSError: [Errno 17] File exists\n\nThis is likely due to changes in the nodepool infra image which removed\nthe default installation of virtualenv.\n\nThis change avoids running virtualenv on an existing virtualenv.\n\nIt also installs the python six package in CI, which is required by\nTenks.\n\nThis commit also enables seed container image build to work around a\nbroken bifrost-deploy queens image on Docker Hub.\n\nChange-Id: Id482ae69fc84b106d9d11a9a1693d94ce5f34ba5\n(cherry picked from commit 050042ceccff5d87ba67e22f36f6c6ce8df7a663)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d9cc2f22535134aece35c731ede7373f06682bf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/8d9cc2f22535134aece35c731ede7373f06682bf"}]},"branch":"refs/heads/stable/rocky"},"5f811269ea9ad23bd82ea620b9fcdd35c5fe7f0d":{"kind":"REWORK","_number":3,"created":"2020-06-18 09:39:45.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/736278/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/736278/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/736278/3"}}},"commit":{"parents":[{"commit":"5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59","subject":"Use ensure-docker role instead of install-docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-16 13:53:06.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-18 09:39:31.000000000","tz":120},"subject":"CI: don\u0027t run virtualenv on an existing virtualenv","message":"CI: don\u0027t run virtualenv on an existing virtualenv\n\nThe seed upgrade job is currently failing on CentOS 7:\n\nTraceback (most recent call last):\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 2327, in \u003cmodule\u003e\n    main()\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 712, in main\n    symlink\u003doptions.symlink)\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 924, in create_environment\n    site_packages\u003dsite_packages, clear\u003dclear, symlink\u003dsymlink))\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 1369, in install_python\n    os.symlink(py_executable_base, full_pth)\nOSError: [Errno 17] File exists\n\nThis is likely due to changes in the nodepool infra image which removed\nthe default installation of virtualenv.\n\nThis change avoids running virtualenv on an existing virtualenv.\n\nIt also installs the python six package in CI, which is required by\nTenks.\n\nThis commit also enables seed container image build to work around a\nbroken bifrost-deploy queens image on Docker Hub.\n\nChange-Id: Id482ae69fc84b106d9d11a9a1693d94ce5f34ba5\n(cherry picked from commit 050042ceccff5d87ba67e22f36f6c6ce8df7a663)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f811269ea9ad23bd82ea620b9fcdd35c5fe7f0d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5f811269ea9ad23bd82ea620b9fcdd35c5fe7f0d"}]},"branch":"refs/heads/stable/rocky"},"b42c5dbf7141826b4c78ab5f8404fd4ce37127dd":{"kind":"REWORK","_number":4,"created":"2020-06-18 10:02:46.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/736278/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/736278/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/736278/4"}}},"commit":{"parents":[{"commit":"5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59","subject":"Use ensure-docker role instead of install-docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-16 13:53:06.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-18 10:01:58.000000000","tz":120},"subject":"CI: don\u0027t run virtualenv on an existing virtualenv","message":"CI: don\u0027t run virtualenv on an existing virtualenv\n\nThe seed upgrade job is currently failing on CentOS 7:\n\nTraceback (most recent call last):\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 2327, in \u003cmodule\u003e\n    main()\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 712, in main\n    symlink\u003doptions.symlink)\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 924, in create_environment\n    site_packages\u003dsite_packages, clear\u003dclear, symlink\u003dsymlink))\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 1369, in install_python\n    os.symlink(py_executable_base, full_pth)\nOSError: [Errno 17] File exists\n\nThis is likely due to changes in the nodepool infra image which removed\nthe default installation of virtualenv.\n\nThis change avoids running virtualenv on an existing virtualenv.\n\nIt also installs the python six package in CI, which is required by\nTenks.\n\nThis commit also marks several jobs as non-voting to allow this fix to\nbe merged. These jobs are broken due to unrelated issues.\n\nChange-Id: Id482ae69fc84b106d9d11a9a1693d94ce5f34ba5\n(cherry picked from commit 050042ceccff5d87ba67e22f36f6c6ce8df7a663)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b42c5dbf7141826b4c78ab5f8404fd4ce37127dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/b42c5dbf7141826b4c78ab5f8404fd4ce37127dd"}]},"branch":"refs/heads/stable/rocky"},"6dd11cea73e0e2df4d65ecc3744e4abc68ff195f":{"kind":"REWORK","_number":5,"created":"2020-06-18 10:05:36.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/78/736278/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/736278/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736278/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/736278/5"}}},"commit":{"parents":[{"commit":"5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59","subject":"Use ensure-docker role instead of install-docker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/5c17d14edd7bae5e0305e54ff3b82f8c50e2dc59"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-16 13:53:06.000000000","tz":60},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-06-18 10:05:30.000000000","tz":120},"subject":"CI: don\u0027t run virtualenv on an existing virtualenv","message":"CI: don\u0027t run virtualenv on an existing virtualenv\n\nThe seed upgrade job is currently failing on CentOS 7:\n\nTraceback (most recent call last):\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 2327, in \u003cmodule\u003e\n    main()\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 712, in main\n    symlink\u003doptions.symlink)\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 924, in create_environment\n    site_packages\u003dsite_packages, clear\u003dclear, symlink\u003dsymlink))\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 1369, in install_python\n    os.symlink(py_executable_base, full_pth)\nOSError: [Errno 17] File exists\n\nThis is likely due to changes in the nodepool infra image which removed\nthe default installation of virtualenv.\n\nThis change avoids running virtualenv on an existing virtualenv.\n\nIt also installs the python six package in CI, which is required by\nTenks.\n\nThis commit also marks several jobs as non-voting to allow this fix to\nbe merged. These jobs are broken due to unrelated issues.\n\nChange-Id: Id482ae69fc84b106d9d11a9a1693d94ce5f34ba5\n(cherry picked from commit 050042ceccff5d87ba67e22f36f6c6ce8df7a663)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6dd11cea73e0e2df4d65ecc3744e4abc68ff195f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6dd11cea73e0e2df4d65ecc3744e4abc68ff195f"}]},"branch":"refs/heads/stable/rocky"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
