)]}'
{"id":"openstack%2Fkayobe~736280","triplet_id":"openstack%2Fkayobe~stable%2Fstein~Id482ae69fc84b106d9d11a9a1693d94ce5f34ba5","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"Id482ae69fc84b106d9d11a9a1693d94ce5f34ba5","subject":"CI: don\u0027t run virtualenv on an existing virtualenv","status":"MERGED","created":"2020-06-17 15:52:51.000000000","updated":"2020-06-18 13:22:38.000000000","submitted":"2020-06-18 13:20:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736280-1592486401690-543b82f4","meta_rev_id":"50707bbcdc127ae9dd871f42e39bc1f77571048b","_number":736280,"virtual_id_number":736280,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 13:20:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-18 11:13:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-18 11:13:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-17 17:53:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 13:20:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"320c42f9883f3966b5945b1fbb0c80c6e2bad7c6","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 15:52:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e5d971743f5f7b362be48c3cce89bc31a387f632","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 17:53:30.000000000","message":"Patch Set 1: Review-Priority+2","accounts_in_message":[],"_revision_number":1},{"id":"ab8e75dc17c4902490fb3b63fde47b60e64b7e8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 17:53:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e539681db4341e18f993cd20a0a7c24 : SUCCESS in 8m 30s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/08b78c08560343cfa608419fa5b5b500 : SUCCESS in 9m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ad453460cc814ae2bd70dadbd2671cf1 : SUCCESS in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a5db2cd1f47412c8d4e00d7d43a53e8 : SUCCESS in 6m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/e921a7cdbc404c16b7dca499639ac4b3 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5882e89f5144086a158346fd44a8c2b : SUCCESS in 6m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9683a143b06c4d89975cade4940ec843 : SUCCESS in 7m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ad4baf8998334c4f9d9edc7a129e3c3b : SUCCESS in 11m 52s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/eb9bcf13f4eb48d4b88186db983eca2f : SUCCESS in 35m 28s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/4466be191e454ccfa165b3e07149689e : SUCCESS in 1h 13m 58s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/dfa0e7a978864b928a6a495e2528ef18 : SUCCESS in 1h 48m 22s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/e3045097b5d44474b9b1b7b6d019cf03 : SUCCESS in 25m 19s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7a498feae8f346ea99fded088b9b1953 : SUCCESS in 38m 26s","accounts_in_message":[],"_revision_number":1},{"id":"da6b1019096a38c03792490ed82a3f4bf9d4d622","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 11:13:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8b7a22866acc209ad263ea3eda742798dac6d2e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 11:14:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"615b76208893d3ed379526d628a5da74acc8a930","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 13:20:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/a175d683be62475dacaefab1adce31b2 : SUCCESS in 8m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2c72854a3eb0468abdcf371df8c62898 : SUCCESS in 7m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b2a079743d094cc7b8e76127c7a5a26d : SUCCESS in 4m 44s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/38d2846e9eec41569a2049b51c1bab46 : SUCCESS in 7m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8428a9195874ac184a64b6521361087 : SUCCESS in 6m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e658a2b83a9e4bcea0b6879e181bc5b0 : SUCCESS in 8m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/27892f6724f44914b60173fa2ae3f5c7 : SUCCESS in 10m 05s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be81fc6847a24aad95b6c67ebc5634e5 : SUCCESS in 37m 04s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/9b310e50503b4c88a33bd7367ee8b897 : SUCCESS in 1h 08m 29s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/d4a2eab1245b42afaa1ab1f9b4b53279 : SUCCESS in 2h 02m 18s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/23ea778ee24b4a95a36bb123d49ff243 : SUCCESS in 25m 36s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/518ccbf598264b13adfef72e1efd2d93 : SUCCESS in 33m 12s","accounts_in_message":[],"_revision_number":1},{"id":"7cf290bf95368ba8898fa52945b33f16036e2c4d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 13:20:01.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"50707bbcdc127ae9dd871f42e39bc1f77571048b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 13:22:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df22acb7d1f044a69450d68de5bcc793 : SUCCESS in 2m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3d22d2c556bd9394ddc9804687a6c0258e8d244b","revisions":{"3d22d2c556bd9394ddc9804687a6c0258e8d244b":{"kind":"REWORK","_number":1,"created":"2020-06-17 15:52:51.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/80/736280/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/736280/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/736280/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/736280/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/736280/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/736280/1"}}},"commit":{"parents":[{"commit":"caa9b3f5e86294be86ce92e60c83f518932a1fa8","subject":"Add missing colon","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/caa9b3f5e86294be86ce92e60c83f518932a1fa8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-16 13:53:06.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-17 15:52:34.000000000","tz":60},"subject":"CI: don\u0027t run virtualenv on an existing virtualenv","message":"CI: don\u0027t run virtualenv on an existing virtualenv\n\nThe seed upgrade job is currently failing on CentOS 7:\n\nTraceback (most recent call last):\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 2327, in \u003cmodule\u003e\n    main()\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 712, in main\n    symlink\u003doptions.symlink)\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 924, in create_environment\n    site_packages\u003dsite_packages, clear\u003dclear, symlink\u003dsymlink))\n  File \"/usr/lib/python2.7/site-packages/virtualenv.py\", line 1369, in install_python\n    os.symlink(py_executable_base, full_pth)\nOSError: [Errno 17] File exists\n\nThis is likely due to changes in the nodepool infra image which removed\nthe default installation of virtualenv.\n\nThis change avoids running virtualenv on an existing virtualenv.\n\nIt also installs the python six package in CI, which is required by\nTenks.\n\nDepends-On: https://review.opendev.org/736278\n\nChange-Id: Id482ae69fc84b106d9d11a9a1693d94ce5f34ba5\n(cherry picked from commit 050042ceccff5d87ba67e22f36f6c6ce8df7a663)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d22d2c556bd9394ddc9804687a6c0258e8d244b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/3d22d2c556bd9394ddc9804687a6c0258e8d244b"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
