)]}'
{"id":"openstack%2Fkayobe~736319","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ia81e70e45c3e8a68720535b4f4d117f9951ed360","project":"openstack/kayobe","branch":"stable/ussuri","topic":"fix-recursive-var-expansion-stable/ussuri","hashtags":[],"change_id":"Ia81e70e45c3e8a68720535b4f4d117f9951ed360","subject":"Fix kolla-ansible config when a network is named \u0027external\u0027","status":"MERGED","created":"2020-06-17 17:56:38.000000000","updated":"2020-06-18 15:12:47.000000000","submitted":"2020-06-18 15:08:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736319-1592492931058-93901803","meta_rev_id":"8b4149ef303b7768d84152c76f0dbab196545ffa","_number":736319,"virtual_id_number":736319,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 15:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-18 08:21:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-17 20:34:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-18 08:21:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":1,"date":"2020-06-17 17:56:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-17 20:34:13.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-18 15:08:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6bb51bf0d56b5f82313f2ed6aff73e00bfc4c85d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 17:56:38.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d0551208c618bffac88b2b57a3cccb66774f9bf7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-17 17:56:46.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"abd7704419703443b9787b1e8ddb98e6ea57e0d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-17 20:00:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76a557e8f8c74322b37a23233f2303f3 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35007a6bbf894713a0ac7d905d5d41bf : SUCCESS in 4m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fd79df2f0ced4ebd96072cf240fadca1 : SUCCESS in 4m 19s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/1bd0ee60d176447faf3243b72f85888c : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e35ce8fd7f542688914bca50f5fd42d : SUCCESS in 8m 15s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/849a2d822ac043afb20a3d43625f1a13 : SUCCESS in 6m 09s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/d6f0c7a9a82b490aabc22351c7212c42 : SUCCESS in 9m 09s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/9ca00d01f87a490392d269d86e6a636e : SUCCESS in 33m 22s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4bd5e3e993264d89ab2de5e9b0d61816 : SUCCESS in 1h 19m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/4776a752dc1941eba9a2e7138b3c7a5f : SUCCESS in 16m 50s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b6e39ae1b92e4c078037243d60ce820d : SUCCESS in 2h 00m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1dd23b95d7214bd5ba029bcde306e314 : SUCCESS in 18m 07s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/910e614e1f714846b510e56c97e1937f : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":1},{"id":"caac132b875b3c88a7a313ab4dc79c4c04b8ea6b","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-17 20:34:13.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4aacef7a1730c0c267c763b5bb8fd31f01310583","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 08:21:33.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4df948fe815c72f79d376c1a1170896138318d9c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 08:21:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"42d8e75096efb7b27079a471b012144bb1ff475d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:26:05.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d093a1422854170bd076e3b1b8a78f0 : SUCCESS in 7m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7e00c64c6b6a46279b9b229c5cb1a9ab : SUCCESS in 5m 43s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a2fa638f833d4608a4a39445ee1ecbfb : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a9712e764734ecf9bcbc6a3c79252dd : SUCCESS in 10m 05s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/69da9ffd0f584420abe5eb8ec8ec1661 : SUCCESS in 5m 50s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6cb53766336e4e2bbd2736ed7d48e526 : SUCCESS in 10m 28s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/90ca8f37db7c4209acc8bbd5922391c9 : SUCCESS in 26m 06s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5c9af8e7521c46429e9aa92fa50c5e5f : SUCCESS in 1h 19m 15s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/06a3dae4fa584cf88d6402d50080eee9 : SUCCESS in 25m 37s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cedc8fb6a3c6484998092f29bdc1653c : FAILURE in 1h 58m 52s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/694fd75a6fb2482eb02e1b5755445a12 : SUCCESS in 18m 32s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fe132867bdad4653bb2f690cefb34b71 : SUCCESS in 29m 18s","accounts_in_message":[],"_revision_number":1},{"id":"1bafd6f73233847655a6770aa4c804cc6c07ac83","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 10:42:44.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"6c509e444069862c9128ef61519530d8fea05cc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 13:02:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c39d64db9bb4793a4c7387230500de6 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8276a39b653349d3a7adde0200a6606b : SUCCESS in 6m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d0db686f5cf4fb59ee1aa5067c3ed09 : SUCCESS in 6m 22s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3abd05b509804218a293c82841bbdaff : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5654e05d6ae1410290c292cad0e9e762 : SUCCESS in 9m 39s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9b04852228344cdd8a29a2b8e9058657 : SUCCESS in 8m 12s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8f13a30e0a8b4538ac3b9d8bab074018 : SUCCESS in 12m 33s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa82898ad49e43bea04d44428414a6ee : SUCCESS in 34m 46s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/c69edb36d5964de88616321423320f87 : SUCCESS in 1h 29m 24s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9044bac6c62d40ca95e5415796e8ab5a : SUCCESS in 18m 39s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/24ff33cf3b584993b9fc641f47bf96f8 : SUCCESS in 2h 17m 32s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a3193476c9b942e1ab06b8ee3fea85a3 : SUCCESS in 20m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/150eef9c7c324f458d34f0ef5b7d77f9 : SUCCESS in 32m 53s","accounts_in_message":[],"_revision_number":1},{"id":"fe5f6add43cfb4a80b2ef19c1066919e6a45c2c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 13:02:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2566bf536bf7782c39be02fa0d5a9d2946d8f252","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:08:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/781942ba7f134a368eea8e70f94957d5 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/209835f0ee2a4369a4e113e8bd5fd18f : SUCCESS in 5m 12s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d170fa7932cf4548a84111fe91e395e7 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24b70ed09142412b9c023f7cd9b6ff2e : SUCCESS in 12m 11s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/06d8c098b7334ab1a762dbbfe7f2a60d : SUCCESS in 9m 25s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/af7b868460124f078cce0de8e0cae9b3 : SUCCESS in 10m 17s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/f5525302dcd54c17b3fffdd2a74bd9f7 : SUCCESS in 35m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7ea9a5c092a24da0b4c8ad037dce507a : SUCCESS in 1h 07m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1ecb7bfa67704f5aaf26be9b22edc801 : SUCCESS in 17m 42s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/31c64ab46d9647d39322d4f351a9af4d : SUCCESS in 2h 01m 45s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/81491a9c94564decb64f78147f1ed5fd : SUCCESS in 18m 48s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8d553181732644deb020511d3dc43447 : SUCCESS in 29m 35s","accounts_in_message":[],"_revision_number":1},{"id":"17ff44f06074cc45249a112e7f31ad237b3fffef","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:08:51.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7772eb80196cb595d6165fdb5e95d30bf144667b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:12:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ef1a3bc021f4d479e63135b3e859752 : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6d3c6c3fde139afded04c19b351c92b35be33ebf","revisions":{"6d3c6c3fde139afded04c19b351c92b35be33ebf":{"kind":"REWORK","_number":1,"created":"2020-06-17 17:56:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/19/736319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/19/736319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/736319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/736319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/19/736319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/19/736319/1"}}},"commit":{"parents":[{"commit":"9646b990ca6fc88a6b254a4ce2a5168c28ce27c2","subject":"Add missing colon","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9646b990ca6fc88a6b254a4ce2a5168c28ce27c2"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-15 09:56:44.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-17 17:56:38.000000000","tz":0},"subject":"Fix kolla-ansible config when a network is named \u0027external\u0027","message":"Fix kolla-ansible config when a network is named \u0027external\u0027\n\nSince I2172a74f4719605f6ec81fadec46ce49f8310a96, kolla-ansible.yml fails\nif there is a network named external:\n\n    recursive loop detected in template string: {{ item in\n    (network_interfaces | net_select_bridges | map(\u0027net_interface\u0027)) }}\n\nRename the task variable \u0027external_interface\u0027 to avoid this conflict.\n\nAffects master only.\n\nChange-Id: Ia81e70e45c3e8a68720535b4f4d117f9951ed360\nStory: 2007814\nTask: 40087\n(cherry picked from commit 47c675c96dc717d552bd3e9bbdbbd0648d7c593d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d3c6c3fde139afded04c19b351c92b35be33ebf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6d3c6c3fde139afded04c19b351c92b35be33ebf"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
