)]}'
{"id":"openstack%2Fkayobe~736478","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ideada0f299f34d16d1286633b6e21404b736918b","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"Ideada0f299f34d16d1286633b6e21404b736918b","subject":"Fix ironic inspector store endpoint configuration","status":"MERGED","created":"2020-06-18 07:51:04.000000000","updated":"2020-06-19 01:18:46.000000000","submitted":"2020-06-19 01:17:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736478-1592529422025-a51fd944","meta_rev_id":"7c48e6e8544d4ec9d8226e856b8d525b42f0f68f","_number":736478,"virtual_id_number":736478,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 01:17:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2020-06-18 21:23:39.000000000","_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"all":[{"value":2,"date":"2020-06-18 07:59:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":2,"date":"2020-06-18 13:35:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":1,"date":"2020-06-18 13:35:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 07:59:28.000000000","updated_by":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"reviewer":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"state":"REVIEWER"},{"updated":"2020-06-18 21:23:39.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-19 01:17:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f48f42cd6deebd74117c4ff32684e9dac3e81ee0","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:51:04.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7bb8626d54f431ec40b247ec2bbc458175b2ff46","author":{"_account_id":17669,"name":"Doug Szumski","email":"doug@stackhpc.com","username":"DougSzumski"},"date":"2020-06-18 07:59:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b21e118a1fd5266660185c843674b7e8a6bbbe5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:36:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df4030a4f9cf4930bc9b04fc1f19c518 : SUCCESS in 7m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0da7be3a29e94030a19e94ab6f3ff810 : SUCCESS in 9m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3c3ec8d48b454981a7aff216355f4aa2 : SUCCESS in 9m 33s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3ce1c5b063104774b2dae8b626e7ea05 : SUCCESS in 9m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c56bf1669ad049cea7ce29dcf9b499bd : SUCCESS in 13m 20s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2328826182324b9391e8332fde74ce43 : SUCCESS in 7m 55s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cb46ae327e8e4b859a244aa27896259e : SUCCESS in 11m 16s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/fa7d61fc176948539fda3614e3918e4f : SUCCESS in 36m 40s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/7af397937c524c46bb5cb3cb84b24dd9 : SUCCESS in 1h 15m 57s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/e89d4a2d86c6443d8fcaa9b6db6234ec : SUCCESS in 16m 43s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b6170dcce5924c64845b15d026716aae : TIMED_OUT in 2h 35m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/1e835527f8df4d45be4ee688bd0ed997 : SUCCESS in 20m 44s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/cc8638786212493cb962c761f93cc4e7 : SUCCESS in 32m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6b022cfb7ceb6e1d12047050893842943dbe0d05","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 10:41:16.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e25716652c6247a4681cf9b387b3e8758e90eba6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:48:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7b262a516fcc4255a73c23ee91eab705 : SUCCESS in 8m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11dce84d7dd44749be7338706113942d : SUCCESS in 5m 48s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5c86383794cc44a289523b27ebf286f3 : SUCCESS in 5m 40s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/6c2c26a672974600951ee64d6aa9e46d : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c208a251aaf8409a9426c11e2ce45a91 : SUCCESS in 9m 17s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/b88fe77c213a4f23a7c982ab08811380 : SUCCESS in 6m 10s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/8c8ec0ac8ff1493b98ea147f99019539 : SUCCESS in 11m 47s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/ec86776d96fc4a818dee44e99ac564f8 : SUCCESS in 37m 32s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/713820aea7974e37836f8793a2957d11 : SUCCESS in 1h 04m 16s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/3e0b563e019441daa643b9f4b3a0dc73 : SUCCESS in 18m 25s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/44a698a3779c472fa2d84b3d9095f2dc : SUCCESS in 2h 05m 30s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/bfa3838c2c0f4eeb8bd304438a768624 : SUCCESS in 24m 03s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6a236ad8d9f3473eb9d12541c680012c : SUCCESS in 40m 08s","accounts_in_message":[],"_revision_number":1},{"id":"360d58618ea2a0176a1f9da1d13e34ae2a257bf7","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 13:35:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c7afbdb5d2ad14f0018265addfd4719b143136e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 13:38:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"93038520902aea7d45f5a69ce4d60cc1381320f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 15:39:15.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2455c6c8fc34d44909023fd7bb6f210 : SUCCESS in 8m 22s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3996675bf0c543f0aad52efd0b37f1fe : SUCCESS in 6m 48s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/354a8e5c287e415792307ab023e3faee : SUCCESS in 7m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f41d829b82424273a537481f9d45db22 : SUCCESS in 10m 13s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/9a9bbe211a244a5bad1b194942e8b08c : SUCCESS in 9m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0d051cb88b4845dc9ba6cafd54f96676 : SUCCESS in 10m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/3670640a935448beb82d4966ea66ee15 : SUCCESS in 37m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/41e3b56a8bf74d0a963adb85bb227109 : SUCCESS in 1h 16m 43s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/16203cbcef2e400b8ccf887660eeec70 : SUCCESS in 19m 06s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/74d44d7f012d44db95c9bf6f27a5e4b5 : FAILURE in 1h 55m 06s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/298e09d8bde64b88b23006080bb33c76 : SUCCESS in 20m 25s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/8ad85009763b4ffbbde073fec1ccd092 : SUCCESS in 27m 18s","accounts_in_message":[],"_revision_number":1},{"id":"03585b78805162d377c7efc2e9d463b4894dca64","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 21:23:39.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b6106c3a2c1de688ca397ea23697680ab27034fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 23:27:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96a4027879a447fa835da445a1739ce0 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8682fc5be7f4ebcb9c98bd2b60d8c17 : SUCCESS in 7m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/84f973b3200a4f048926373091ea73e5 : SUCCESS in 7m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3bce1b0c7070445db1129fd7e686c74d : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/700f57500597482393a15a0eb714df57 : SUCCESS in 6m 45s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/5bfe78b54f5946b38cc22cd2e96c69a3 : SUCCESS in 4m 31s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/2006ed6c677d4338bbcfef96d9ad173e : SUCCESS in 10m 32s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/e170eaadb0504cc9bdd68aab58012a6b : SUCCESS in 35m 59s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/5a8540b49c8f4795a68ea16f0b5da240 : SUCCESS in 1h 12m 06s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/0cd1be219f1e406ca88cc35453a66944 : SUCCESS in 15m 59s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f7e94a801ebe4e1d80d89320fb8b9da2 : SUCCESS in 1h 58m 19s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/a072698415234f71917b10b413791d5b : SUCCESS in 18m 33s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/f1a13d64f3eb46f2a90f7e1f484455d1 : SUCCESS in 39m 20s","accounts_in_message":[],"_revision_number":1},{"id":"38d0fc3aa0904087a45a8d945c04a7c5c779fd0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 23:27:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e3d2b35475ee0cc8439e80aa56b6eeacc03a4058","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 01:17:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23a4dcb15e324c1ea5178ded3595e631 : SUCCESS in 4m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b09df033de440a78f6096bcb4f00781 : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/d3ab750d59274dd685531f9cf0e0dc09 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c64caa23c88245fa91844ec5403e06ce : SUCCESS in 7m 58s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a704356850b64e35a995f3c0cf61d761 : SUCCESS in 4m 35s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f61b3e72c124a60bc8696616b2b3524 : SUCCESS in 8m 43s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/be0f35ef2a4b4636adc86077717d280e : SUCCESS in 26m 59s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a0531295ebd94cd682a612c21eb6be27 : SUCCESS in 1h 05m 00s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/183aabaf02d94b2085e923f4c2e44542 : SUCCESS in 17m 35s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/1bfc694fd6214573b3472032106c4b4f : SUCCESS in 1h 48m 01s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/05eda46e2f6a44709e57ba5f6c3e5491 : SUCCESS in 20m 28s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/ef43e58a79394f3288a8094d98bf7396 : SUCCESS in 26m 57s","accounts_in_message":[],"_revision_number":1},{"id":"ff0ffdf2ac2c69f33d8d423137733845862b1e7b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 01:17:02.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"7c48e6e8544d4ec9d8226e856b8d525b42f0f68f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 01:18:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1f6c3dbe83a49198be699461954d03c : SUCCESS in 1m 21s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ba8413814342afdb436f62aed4912a32d9bc0c7","revisions":{"9ba8413814342afdb436f62aed4912a32d9bc0c7":{"kind":"REWORK","_number":1,"created":"2020-06-18 07:51:04.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/78/736478/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/78/736478/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736478/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736478/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/78/736478/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/78/736478/1"}}},"commit":{"parents":[{"commit":"e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e","subject":"Merge \"Add framework to deploy user-defined containers on seed\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 14:12:36.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 07:51:04.000000000","tz":0},"subject":"Fix ironic inspector store endpoint configuration","message":"Fix ironic inspector store endpoint configuration\n\nWhen performing inspection on baremetal compute nodes, inspector\ncurrently has the following failure when storing inspection data to the\nintrospection store:\n\n    No \u0027keystoneauth1.plugin\u0027 driver found, looking for \u0027token_endpoint\u0027\n\nThis happens because the token_endpoint auth type is no longer\navailable. The new way to do this is to set auth_type \u003d none and provide\nan endpoint override.\n\nStory: 2007795\nTask: 40049\n\nChange-Id: Ideada0f299f34d16d1286633b6e21404b736918b\n(cherry picked from commit f0d1b630fc960923717fb0b3e85f56935298fd4d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba8413814342afdb436f62aed4912a32d9bc0c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/9ba8413814342afdb436f62aed4912a32d9bc0c7"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
