)]}'
{"id":"openstack%2Fkayobe~736480","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I4a72d9ae49ad41716522c3074c16d8ca23c3ff94","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I4a72d9ae49ad41716522c3074c16d8ca23c3ff94","subject":"CI: Update IPA images during upgrade","status":"MERGED","created":"2020-06-18 07:52:07.000000000","updated":"2020-06-19 13:11:14.000000000","submitted":"2020-06-19 13:08:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736480-1592572117159-85deaf72","meta_rev_id":"6c49d2f5d7462948d7ec21da4b02e0cb123597f3","_number":736480,"virtual_id_number":736480,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 13:08:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-19 10:39:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-19 10:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 10:48:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 10:48:27.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-19 13:08:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e206bb72cd04ed784e301666dd4dd2aa775f04a2","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:52:07.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"fb28243afdb09378c3d894c3534782f4e1d10124","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:07:10.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5349b11c6534804899f0aecde1a5634 : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6b18fa36f4649d18e0e05f556238d58 : SUCCESS in 10m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a8acaac00e1b453f92038c268c09839d : SUCCESS in 8m 46s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38231be19a4a49bfb4b96ee426a9471f : SUCCESS in 9m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a4f8f888674794b7efc59f0c2db926 : SUCCESS in 10m 26s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/402e32248e74452793fe334aad50fea2 : SUCCESS in 7m 29s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7f9d3ea32edf42a08fc232eebeddc0d8 : SUCCESS in 13m 31s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/7744e2b6923a417bb4f4358b9fbb6d58 : SUCCESS in 29m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/ae3d009377a74f9895c6f438479b59c0 : SUCCESS in 1h 18m 49s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/a683ad6dcd774a63b213d263bbcfef77 : SUCCESS in 19m 33s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/a7e92cc099f346f3af73d5597fd09c42 : FAILURE in 2h 01m 26s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fc623e73d66e45f38c05ffbc114086d4 : SUCCESS in 19m 42s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/4fccd678f6984ee780baf168b6d0e8b7 : SUCCESS in 29m 42s","accounts_in_message":[],"_revision_number":1},{"id":"31a3f9e2d22dcd95f997eb3ff6f36af50f643f8d","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 12:43:50.000000000","message":"Patch Set 1:\n\nrecheck\n\n504 when deleting baremetal instance.","accounts_in_message":[],"_revision_number":1},{"id":"088aba3727117f2a10bf158f2daa95930d009813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 14:39:46.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52c206c42ab24575a90e3abe89d2670c : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8eb5a35f5dc439491f4015fea023264 : SUCCESS in 5m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3e611cccbccf4a7bae09a4269daf50b2 : SUCCESS in 8m 14s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/3a5fe20240c44a0c84df4f209da8b8bd : SUCCESS in 7m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9b9a886500f4fdda551903e0d6ed94d : SUCCESS in 16m 27s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2b6124de47854f8289a4a797fa00d7f2 : SUCCESS in 9m 26s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b0aeb527bd72433b93cf419f5fbf5ab3 : SUCCESS in 10m 04s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0b7c877321164cfda76babead776c9ae : SUCCESS in 29m 50s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/07bd551ee3434f7e98b1751a3ecb6eac : SUCCESS in 1h 08m 33s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/8fce5249bf33410daad380b32d4a0bec : SUCCESS in 16m 09s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/48a7c72019af476d99b33b2be8b04e98 : FAILURE in 1h 47m 59s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2db8e059137041cc8b909d48f3607aef : SUCCESS in 20m 31s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/e231bfe4c90349f39a19c1434aab86b5 : SUCCESS in 21m 54s","accounts_in_message":[],"_revision_number":1},{"id":"95c741d241c49f043604c41c4233e3ae4227309a","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 16:17:59.000000000","message":"Patch Set 1:\n\nrecheck\n\nout of disk space","accounts_in_message":[],"_revision_number":1},{"id":"de86e6bab296cde96093c6e7a38f6a804bf6da00","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 18:25:06.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a78678ff3f6b4e6ebe7722853ff9aa95 : SUCCESS in 9m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/50516a5073cf40d09889cce2e64176c7 : SUCCESS in 6m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ad2cdb4d5fd4ee4a9d1db60b0e6dc74 : SUCCESS in 5m 36s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/34da0c7ad892433fb1ab0ba2db67fe05 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cada6dd773094b34bcc62e9153947660 : SUCCESS in 7m 51s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/6569a89f206441509a1071adeccfa56f : SUCCESS in 6m 33s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/e61c766b87e84b139130c52eb71141fe : SUCCESS in 8m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8162f67790c448c1a90c448b1a5719b2 : SUCCESS in 37m 07s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/eef4226da10748388a0fb1a314d80e40 : SUCCESS in 1h 12m 17s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/07c7c6cbc1404e39955c203f8d942ced : SUCCESS in 18m 07s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/16ce5f35a4e44abf954f445f03a96c17 : FAILURE in 1h 57m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/7d72a12612774784b858ed40f0d302f4 : SUCCESS in 21m 50s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b87217ef33884af5a2750bf014421797 : SUCCESS in 28m 12s","accounts_in_message":[],"_revision_number":1},{"id":"0e965d50bb97d32bfabd579534a24f1e84a6b0dd","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 19:36:56.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"0b97911a649f6e42349caa3f467e1332508c60b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 22:03:48.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a7782a50f97241d786393daea0117b8c : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1468afd33ee4d30b69c4d6b7ef6e60f : SUCCESS in 5m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6a4f9ce827104bffb6af9f191d7bfb5c : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/8dd8413c090847f3aa3e13f1d26cbaf0 : SUCCESS in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc61115d3da1478c89ce7f33d11c4729 : SUCCESS in 6m 12s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/00ea7a26e32c4657bb944d05a40c4666 : SUCCESS in 6m 46s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/1ebded6ee44442d2afa23e1086accf32 : SUCCESS in 12m 36s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/1dc71ebeb8fc47389a4ab26867532e73 : SUCCESS in 35m 15s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/e7625cd1c4ee429992e56219bf41ea3a : SUCCESS in 1h 20m 58s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2a7464fd71874788b47367e148d193ba : SUCCESS in 16m 24s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/9dc9bde0c58c40569411f0511867e05a : FAILURE in 2h 21m 44s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/ef723fcbb95c4f999fb87e0ceabce019 : SUCCESS in 17m 41s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/235e8db8109f4823b355a2796d69a2f2 : SUCCESS in 29m 46s","accounts_in_message":[],"_revision_number":1},{"id":"b3ed480d46d74214bc77fc385c523c8edb97723e","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 07:24:49.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"dd7221a6cfc6522d4356b3e6100823651a4f076c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:25:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc90a4ab70e649c6937f90c63b9cdcd9 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce992ab02f43459b9b6f5f639ee9ad46 : SUCCESS in 7m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2573d15f337244a4b8ea6ee6f096a178 : SUCCESS in 5m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/cb8c927164b344199d6bcbd7541407a8 : SUCCESS in 9m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f462e2c0f22f4e9fafdfdfefff44d945 : SUCCESS in 10m 28s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/58dfd34455ae4d2b9bf931940ff29fe9 : SUCCESS in 6m 45s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a79dd68765ae45948306cd435143cfc9 : SUCCESS in 11m 24s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/0c146fb6c29845339773dab985262022 : SUCCESS in 39m 01s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/b74c068cc38a4dffabb2d6b58106f9e4 : SUCCESS in 1h 23m 49s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/1a34f727d9f84941a998c60a2ba8612c : SUCCESS in 18m 44s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7a88c0e6598a462086271580be408a0d : SUCCESS in 1h 54m 29s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/d86379aebe8e4ccfaa640180e65a3265 : SUCCESS in 20m 13s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/5ca90f023a3c449fa6815909f01277c7 : SUCCESS in 34m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8c00ebacf1c496f03111598d727e65f0e8a9f428","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-19 10:39:36.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"e4f6bbf076843da2c49e09804ef2f01c378f1ec6","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 10:48:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f8a3b568443770e0e04f4ac56a4309a1b60c22f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 10:48:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"544a0c1f984a9675f6e1b11b946c9f58939893c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:08:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8207d3b483d7454f837b43199b88cf8a : SUCCESS in 4m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5f201e51047f4bcf94fb09d47fca0524 : SUCCESS in 4m 38s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/38caeb30ee8e4a50a568a5ddf3fff05b : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15c0149a238f400cba72a659c2cb489f : SUCCESS in 7m 04s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/a34e816627634f92a5066a163fc83047 : SUCCESS in 4m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/7a9ca49b633c4cdfba65e8804b9b3152 : SUCCESS in 9m 30s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/38b3dd0441944d29869a205c7886d841 : SUCCESS in 26m 12s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/178eef3a15ed45eaaab3264926799c50 : SUCCESS in 1h 14m 03s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/9a9696449de544f1ab46a327f9d22421 : SUCCESS in 17m 46s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/29f8701f6d734a289ce914806385557d : SUCCESS in 2h 18m 02s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/edca14ffdbdf4f9f8e568b7d248d8c69 : SUCCESS in 19m 43s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/baa5e5b8365c4092a66989a27f8e8975 : SUCCESS in 32m 17s","accounts_in_message":[],"_revision_number":1},{"id":"8109b70ca73bffa8603a960ad5fbd1c4a9c96534","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:08:37.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"6c49d2f5d7462948d7ec21da4b02e0cb123597f3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 13:11:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/576fc2f2a96f4de5851a1ddf26ff6306 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e5570816102dbfd941f0db51db72e4beb2953c1","revisions":{"6e5570816102dbfd941f0db51db72e4beb2953c1":{"kind":"REWORK","_number":1,"created":"2020-06-18 07:52:07.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/80/736480/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/80/736480/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/736480/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/736480/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/80/736480/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/80/736480/1"}}},"commit":{"parents":[{"commit":"e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e","subject":"Merge \"Add framework to deploy user-defined containers on seed\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 15:46:48.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 07:52:07.000000000","tz":0},"subject":"CI: Update IPA images during upgrade","message":"CI: Update IPA images during upgrade\n\nThis ensures we are using the appropriate IPA images in the upgraded\nenvironment.\n\nChange-Id: I4a72d9ae49ad41716522c3074c16d8ca23c3ff94\n(cherry picked from commit 3d9c5861346d1d14e623b302bc6f5aacfa38c12e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e5570816102dbfd941f0db51db72e4beb2953c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6e5570816102dbfd941f0db51db72e4beb2953c1"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
