)]}'
{"id":"openstack%2Fkayobe~736481","triplet_id":"openstack%2Fkayobe~stable%2Ftrain~I4a72d9ae49ad41716522c3074c16d8ca23c3ff94","project":"openstack/kayobe","branch":"stable/train","hashtags":[],"change_id":"I4a72d9ae49ad41716522c3074c16d8ca23c3ff94","subject":"CI: Update IPA images during upgrade","status":"MERGED","created":"2020-06-18 07:52:25.000000000","updated":"2020-06-19 09:46:25.000000000","submitted":"2020-06-19 09:44:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736481-1592559892216-f3c8871e","meta_rev_id":"aa2d7a9c530e0cb0537d776f6245086831705320","_number":736481,"virtual_id_number":736481,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 09:44:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-18 16:42:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-19 07:25:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 07:25:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 07:25:08.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-19 09:44:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"867e8c2c050551eff5b81cd9663732d8efa1b9eb","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:52:25.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"3c14e7e7924aeb6e9f2e5ecbefa9ed8064a6f930","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 09:28:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/635ddc1d092047cd8c63a4179a9985a3 : SUCCESS in 7m 19s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/763ddf918d45456fa59387b82730eacc : SUCCESS in 8m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c108084c05d6413794d75902dde48c04 : SUCCESS in 7m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3872791dfbb04e838e21c067313aec79 : SUCCESS in 6m 37s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f0014db4a69b44b39d37243ee8685352 : SUCCESS in 8m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b844c94230464b659ba43bcc769efca7 : SUCCESS in 10m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/75b09a939cc34280b686ca159681a7ad : SUCCESS in 9m 58s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/b7808eb18c08416a918221c18e143a3d : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/db9d71898736429ea86c8ee821cb8b29 : SUCCESS in 38m 04s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/b80e9460ada34fb087b30c649fe14ccc : FAILURE in 1h 07m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/4c1d6cc342984ad79cf24389f443cb15 : SUCCESS in 1h 18m 54s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/031ca3ababd74368a93a60c96a12ee1a : SUCCESS in 19m 05s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/ed92b175a8b94288b01c1589a153e7e9 : SUCCESS in 18m 01s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/1f409d5472cd4e8e90dcbb79f3067045 : FAILURE in 1h 00m 17s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/4cd14998b5ae486aa90e2b109eb2631b : SUCCESS in 27m 42s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/6b548602ebce412091137b02a1e8a736 : SUCCESS in 19m 45s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/9e056f78e78c4d64acae0c958fe2c5a1 : FAILURE in 24m 44s","accounts_in_message":[],"_revision_number":1},{"id":"a91f939b477a7d2274cb3bee2268f5eb6d053ee4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 14:11:31.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"e1f007db7844dddd5779779273974074ea0c316c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 16:38:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/366556c7a26b46a9ba5a1a6fd2a875d3 : SUCCESS in 4m 58s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b7e4ba3419774be893826dad5b6bddd6 : SUCCESS in 6m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dfae1133904342499eb16f7fda131e1b : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f63edea21e94780ba48a29f9a31a147 : SUCCESS in 4m 33s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/f12ea9cdd0364917ac0c2651042c2274 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73cd0b6f819a430c84c887bbd13ffb89 : SUCCESS in 7m 23s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/be1f021c45c54b408f20f9dbe9e8a881 : SUCCESS in 5m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/ddd991894d2b4390a048138659cdf10c : SUCCESS in 9m 39s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/acee33b2abdf4159b730e032a9559ea2 : SUCCESS in 36m 35s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/ad1c9707fa044540aacaeb894adc2194 : SUCCESS in 1h 26m 02s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/6b0227d686414070abd86cf2c252b252 : SUCCESS in 1h 22m 42s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/11220379bf6d48e8880661bb60255410 : SUCCESS in 17m 53s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/94640d1454eb43159d4793e0880f8485 : SUCCESS in 19m 21s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/e3e71060216f400c8c6b67ff8c7539a4 : SUCCESS in 2h 16m 17s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/8e69c913982744fbbf7ef67f87b82238 : SUCCESS in 25m 54s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/5fe2db6831ba4b7484bd05df1c6cb2cf : SUCCESS in 21m 29s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/4444b5c2ccfb43a188995689a810af38 : SUCCESS in 38m 41s","accounts_in_message":[],"_revision_number":1},{"id":"adab44e30096f0351ee2dafd5c86727846337404","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 16:42:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c9705286b1c6bdc7c61c77cb9119d6511e3cfe4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 07:25:08.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"82e7c9c70a21a57fe7e98eba83311989bec7894c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 07:25:20.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fa5c12eae89dcfb42ec6573bd88ee38c16e74a3c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:44:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/afff9ee9aee34c24a73c6f9005148bf3 : SUCCESS in 6m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/02578ce58f2a485383717ee9e6282d38 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c553ff2fcd274fb5a843b7084c99cb51 : SUCCESS in 6m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b8d7fea66bde481cb5f4db40a852ab41 : SUCCESS in 6m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf909d3f5064a69a5700906e31eb70d : SUCCESS in 6m 53s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/e63f1f4ab8d74f2fa3fd49b658619107 : SUCCESS in 5m 59s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/727532a903374b96be2a62e4696f28fd : SUCCESS in 8m 29s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/66e7768eff814e128d482f7dad3cc92e : SUCCESS in 26m 54s\n- kayobe-overcloud-centos7 https://zuul.opendev.org/t/openstack/build/db888925c7794ee595c54991efbf70a1 : SUCCESS in 1h 16m 48s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/a093e747e812471c8cc4c67d47a727f2 : SUCCESS in 1h 11m 26s\n- kayobe-overcloud-host-configure-centos7 https://zuul.opendev.org/t/openstack/build/d000678310ff4bc2afbd5be55d93b11d : SUCCESS in 18m 07s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/98823aecbc5e46f1a9d4ef1097bb7ab2 : SUCCESS in 20m 11s\n- kayobe-overcloud-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/a9f5a404d2bc4d379c7165957ab39d95 : SUCCESS in 2h 13m 47s\n- kayobe-seed-centos7 https://zuul.opendev.org/t/openstack/build/2609f00f528a4d7e93971e615fc22d6a : SUCCESS in 21m 59s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/2b0a6780a59141f6a1b23863061222d6 : SUCCESS in 18m 18s\n- kayobe-seed-upgrade-centos7 https://zuul.opendev.org/t/openstack/build/6e8b9911ab414583b636ac7485fafde4 : SUCCESS in 40m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c460289b6704f6422fe7d314c91d3bb4647932b3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:44:52.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"aa2d7a9c530e0cb0537d776f6245086831705320","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:46:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8d1b107117147009fdbc8774de6d3a4 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1774d9e92e3f91127e5c974a6a85d8802a0e2e79","revisions":{"1774d9e92e3f91127e5c974a6a85d8802a0e2e79":{"kind":"REWORK","_number":1,"created":"2020-06-18 07:52:25.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/81/736481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/81/736481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/736481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/736481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/81/736481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/81/736481/1"}}},"commit":{"parents":[{"commit":"62c4de4190d98fc2cc7cd91bc6fd24c6293b614e","subject":"Add missing colon","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/62c4de4190d98fc2cc7cd91bc6fd24c6293b614e"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 15:46:48.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 07:52:25.000000000","tz":0},"subject":"CI: Update IPA images during upgrade","message":"CI: Update IPA images during upgrade\n\nThis ensures we are using the appropriate IPA images in the upgraded\nenvironment.\n\nChange-Id: I4a72d9ae49ad41716522c3074c16d8ca23c3ff94\n(cherry picked from commit 3d9c5861346d1d14e623b302bc6f5aacfa38c12e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1774d9e92e3f91127e5c974a6a85d8802a0e2e79"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/1774d9e92e3f91127e5c974a6a85d8802a0e2e79"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
