)]}'
{"id":"openstack%2Fkayobe~736482","triplet_id":"openstack%2Fkayobe~stable%2Fstein~I4a72d9ae49ad41716522c3074c16d8ca23c3ff94","project":"openstack/kayobe","branch":"stable/stein","hashtags":[],"change_id":"I4a72d9ae49ad41716522c3074c16d8ca23c3ff94","subject":"CI: Update IPA images during upgrade","status":"MERGED","created":"2020-06-18 07:52:36.000000000","updated":"2020-06-19 09:46:23.000000000","submitted":"2020-06-19 09:44:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736482-1592559893358-ec6d586d","meta_rev_id":"9d16ea5b3821a38fe61037dc8e6902b4cabeaf11","_number":736482,"virtual_id_number":736482,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-19 09:44:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":2,"date":"2020-06-18 16:16:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-06-19 07:25:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-06-19 07:25:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-19 07:25:14.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-06-19 09:44:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9db0629d3fdc7180d98332691f189eaadd392989","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:52:36.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"8ea5d9298280117f8d96969586998f38e8eaefdc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 09:17:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c319f34c90740699d71f7d5acada434 : SUCCESS in 7m 45s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/b91a0209c7c94ed090c1ad2dc629eb70 : SUCCESS in 7m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fb79cdd15b0b4a2cbcecdd88ebc28d4e : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c192e4866e949cbb268ec7e6f8f3156 : SUCCESS in 5m 32s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/460548f70f12446d9bd33417fc2d1fd0 : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/358583a10a1148aab79273233fa0236d : SUCCESS in 7m 06s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/ae3c858960d247e3998c79ec92cde99c : SUCCESS in 7m 00s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/cce17e08db564304b1f052f48394b35a : SUCCESS in 12m 02s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/414638a0c30c48cca74dd8bea20700a2 : SUCCESS in 34m 55s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2116a3e93f934558aca82bf4ba855b7d : FAILURE in 1h 02m 52s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/405965bad0d14b04af0f4c19b92c620b : FAILURE in 1h 00m 04s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/b5d461406c9246b7a1abffb1ca94b024 : SUCCESS in 27m 40s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/434ce23d34144bf6b86ae8c0f6c73a43 : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":1},{"id":"8c87ab91367ef3d3dc59979f459adb99ecd72834","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-18 14:11:52.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"b3d91bc121d8f3b25b4223ed6fd4dcf7a5b301c7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 16:13:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/033b00b68f074951a90aa49eb124f9d0 : SUCCESS in 5m 43s\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/2b7e30bbf9664b57ac780d6021c1adf5 : SUCCESS in 7m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/47be8f592ed94d2e9df46f8648c68463 : SUCCESS in 4m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26ee8dcde5d4428c9583980bf925090d : SUCCESS in 6m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/92b214adeee1449b88e164d5233975ae : SUCCESS in 6m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e66b505bf2aa403eb4e9127e6af1ebde : SUCCESS in 5m 29s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/47f9d45a1a0c4da68d76a9d71d2bf846 : SUCCESS in 12m 43s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/abaab03bfeb84168854f87bd6a1d1092 : SUCCESS in 11m 42s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/8ff1f875aee740269f174e1c30ca4d32 : SUCCESS in 36m 12s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/2fa213f852dd44879d45580187226396 : SUCCESS in 1h 10m 00s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/e9f640b98d3e4b66b214775866b99588 : SUCCESS in 1h 51m 33s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/02be705b8ef7454b921ab6ad4b107196 : SUCCESS in 21m 38s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/7ec31794a07f4f0185c666060ef6bdf9 : SUCCESS in 43m 44s","accounts_in_message":[],"_revision_number":1},{"id":"27c9c111cd50519092f68b79763a54914653b723","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 16:16:17.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"0d7a55e24279878cd9f5fa3168907690266b0568","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-06-19 07:25:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"5113726db404dc3ea53225d68e0ca68078925cd8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 07:25:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"89ff2b86b782d0fc722617e50858535cd3172855","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:44:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-py35 https://zuul.opendev.org/t/openstack/build/81555f47ead144f9a9a795b1898a8c80 : SUCCESS in 6m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b607a106ad2448b866d770a322e3473 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7e82809361141caa19aaac2a50bcc44 : SUCCESS in 5m 29s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/b22b665c89904799a69cbac16774f317 : SUCCESS in 6m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f40ed062785747a29d21d4a386c7e9a4 : SUCCESS in 8m 00s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/bb113a99353e460db5031aac9f2579a0 : SUCCESS in 9m 28s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/a810761abf7d4b98b161d9dff97d8ede : SUCCESS in 11m 15s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d1f697f0670049ebbcdfbcdc6ee4d926 : SUCCESS in 36m 56s\n- kayobe-overcloud-centos https://zuul.opendev.org/t/openstack/build/deb1e9bfffff43f785d5bb34386a7539 : SUCCESS in 1h 08m 18s\n- kayobe-overcloud-upgrade-centos https://zuul.opendev.org/t/openstack/build/68dc1133af1142f7a18facd14db70bff : SUCCESS in 1h 59m 49s\n- kayobe-seed-centos https://zuul.opendev.org/t/openstack/build/06df5ecd5a1d487cad5147215ca2dc8e : SUCCESS in 23m 27s\n- kayobe-seed-upgrade-centos https://zuul.opendev.org/t/openstack/build/639f9845004c4b4d86c773cfbacab3b2 : SUCCESS in 40m 50s","accounts_in_message":[],"_revision_number":1},{"id":"21e57a8f8648d07bad916c43d2c2cd191915f1d9","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:44:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"9d16ea5b3821a38fe61037dc8e6902b4cabeaf11","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-19 09:46:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1abeab49ba4450f8e128780ce2417ab : SUCCESS in 1m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"044dd31a0fccdc685b7386088e40aec1d4292eae","revisions":{"044dd31a0fccdc685b7386088e40aec1d4292eae":{"kind":"REWORK","_number":1,"created":"2020-06-18 07:52:36.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/82/736482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/82/736482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/736482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/736482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/82/736482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/82/736482/1"}}},"commit":{"parents":[{"commit":"caa9b3f5e86294be86ce92e60c83f518932a1fa8","subject":"Add missing colon","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/caa9b3f5e86294be86ce92e60c83f518932a1fa8"}]}],"author":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-11 15:46:48.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 07:52:36.000000000","tz":0},"subject":"CI: Update IPA images during upgrade","message":"CI: Update IPA images during upgrade\n\nThis ensures we are using the appropriate IPA images in the upgraded\nenvironment.\n\nChange-Id: I4a72d9ae49ad41716522c3074c16d8ca23c3ff94\n(cherry picked from commit 3d9c5861346d1d14e623b302bc6f5aacfa38c12e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/044dd31a0fccdc685b7386088e40aec1d4292eae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/044dd31a0fccdc685b7386088e40aec1d4292eae"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
