)]}'
{"id":"openstack%2Fkayobe~736483","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~I2f8ecad57f97e5cd3e15156959c017ad29335e31","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"I2f8ecad57f97e5cd3e15156959c017ad29335e31","subject":"Update feature flag defaults","status":"MERGED","created":"2020-06-18 07:53:06.000000000","updated":"2020-06-18 12:35:08.000000000","submitted":"2020-06-18 12:35:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736483-1592483708970-5b55ac28","meta_rev_id":"8be62c4a56cc080b6467c19fb8c39342ec75cc76","_number":736483,"virtual_id_number":736483,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 12:35:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-18 08:14:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-06-18 10:23:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-06-18 10:23:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 08:14:16.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-18 12:35:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"36f3a6e0d94a27ef1a4a197c95e1f09958a6fe94","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:53:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5decd4255b94bdccd0aeb01450f53418e75dcc85","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-18 08:14:16.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"867651aea65a887fb883f4579903108b1cbfb88f","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 10:23:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f6d7111358e29d9e5d78c99388bcbb39a70ac79f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:25:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d683d62600b9492db2e0e0bd2ee34489 : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85bf3ac103f448ccac1159cfae9de202 : SUCCESS in 7m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6560eec262974971b0115e3f0b7651c2 : SUCCESS in 6m 51s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/96ba448fcc75443f99b3c4e9ce454706 : SUCCESS in 6m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/249ae30e2111493295b6aea701942c67 : SUCCESS in 9m 56s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/2bd840ed94a147a0a15d50f9e8040e9d : SUCCESS in 8m 42s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/6f03873c3e9a4ad6a4ecc0d17b0d2a76 : SUCCESS in 10m 48s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/4d8d510ccad24ccba654e999cfefc589 : SUCCESS in 37m 03s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/923d20c295e849728519286ff5dc0dbc : SUCCESS in 1h 12m 02s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/164a8a47c38d48919ac50c10f5c2fc4f : SUCCESS in 20m 16s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/fbed88db3bf64a9bae718406569ce774 : SUCCESS in 2h 02m 55s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c756d4645f7a42fb801884a4d70cebf7 : SUCCESS in 17m 58s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/efdafcd8ba3f400488d1da1d66c24246 : SUCCESS in 31m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ec1538ed9266fdf90b93787341994d26667cff3d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:26:10.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c5dd3bb6ca0dce34f47ebeab79dde94ef65d8c12","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:35:08.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"8be62c4a56cc080b6467c19fb8c39342ec75cc76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:35:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9b4dd0edd62479e98e387994e131f4f : SUCCESS in 6m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9edeb1cbd974413c8dd4814fa1b26272 : SUCCESS in 6m 21s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2ee7a46d72094b3bba394184fafbab63 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd54fe4f1c004c349d6a6baf37f644af : SUCCESS in 11m 02s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/547ffe87146a42f5ba8d20acef82de66 : SUCCESS in 5m 57s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/0b17093da57945a19e3459a268fd828f : SUCCESS in 15m 01s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/76e8f83cb4aa4c44aa51064394f491a6 : SUCCESS in 26m 54s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/8e583e9fbe464b96a6b64d76d86c2eeb : SUCCESS in 1h 21m 50s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/039373dad96742e994b2c7144e695617 : SUCCESS in 20m 03s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/7d794e9c56c145ac9d470119ca35dd5a : SUCCESS in 2h 00m 18s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/fac1b4acea0a4edb999fe2b5aa67172d : SUCCESS in 21m 35s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/6ba00b9ddbb143e5af805aa83d6218e7 : SUCCESS in 28m 23s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"842d4b999df4a043e433e85594d725bedbb9b6cb","revisions":{"842d4b999df4a043e433e85594d725bedbb9b6cb":{"kind":"REWORK","_number":1,"created":"2020-06-18 07:53:06.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/83/736483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/83/736483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/736483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/736483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/83/736483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/83/736483/1"}}},"commit":{"parents":[{"commit":"e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e","subject":"Merge \"Add framework to deploy user-defined containers on seed\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-06-12 12:14:51.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 07:53:06.000000000","tz":0},"subject":"Update feature flag defaults","message":"Update feature flag defaults\n\nWe need these defined to figure out which images to build.\n\nOnly flags that are required by build regex map have been added, and\ndependencies of those.\n\nTrivialFix\n\nChange-Id: I2f8ecad57f97e5cd3e15156959c017ad29335e31\n(cherry picked from commit bde426c9b19b272f350cdc2a84a1b324e2ad3c91)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/842d4b999df4a043e433e85594d725bedbb9b6cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/842d4b999df4a043e433e85594d725bedbb9b6cb"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
