)]}'
{"id":"openstack%2Fkayobe~736484","triplet_id":"openstack%2Fkayobe~stable%2Fussuri~Ia0aa15300f58fc4e912dea7ed304a1cdaf51382a","project":"openstack/kayobe","branch":"stable/ussuri","hashtags":[],"change_id":"Ia0aa15300f58fc4e912dea7ed304a1cdaf51382a","subject":"Update image regexes","status":"MERGED","created":"2020-06-18 07:53:23.000000000","updated":"2020-06-18 12:38:57.000000000","submitted":"2020-06-18 12:35:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"736484-1592483709853-d2b572f1","meta_rev_id":"55363c2a15aa703e39e8e2590178a118f6da37d2","_number":736484,"virtual_id_number":736484,"owner":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-06-18 12:35:09.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"all":[{"value":2,"date":"2020-06-18 08:14:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":2,"date":"2020-06-18 10:23:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":1,"date":"2020-06-18 10:23:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},{"value":0,"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-18 08:14:29.000000000","updated_by":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"reviewer":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"state":"REVIEWER"},{"updated":"2020-06-18 12:35:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9af6f1521e90b17b81b60c1f5e70ea18c0d3dd8","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:53:23.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"0e39addc66abea8640ba4252a602d2ec9765c0da","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 07:53:38.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"f6278f0597876a3cb361f110151b32bd9af46f06","author":{"_account_id":28048,"name":"Will Szumski","email":"will@stackhpc.com","username":"jovial"},"date":"2020-06-18 08:14:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"cd1c388ab574f83aa63a89bc5bd9550175fc05a5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:19:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb51df34d61d4cb88764a2ec541db85d : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5bedb858e404d71b65eee8f851b2629 : SUCCESS in 9m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d632df126314ab0b113a4d61114084f : SUCCESS in 5m 52s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/a1046782852142c790e6c6c0e2895c92 : SUCCESS in 7m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56eb40c8005e45c7a26cec7d8a608b58 : SUCCESS in 7m 48s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4a52a4a84a3e4183ad84dd52eb35036f : SUCCESS in 7m 01s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/28ff45dfcb394d68b7598c60ae0ac244 : SUCCESS in 12m 12s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/d3f6fc05c5ec41d79cc95ebf11325043 : SUCCESS in 29m 23s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/377c3e38747347fa972eba4aec759326 : SUCCESS in 1h 09m 36s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/d113babe509148e09078beca04efe536 : SUCCESS in 18m 32s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b85920e8420b46f5895bcdb311adb9e4 : SUCCESS in 1h 57m 13s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/49283ebc42054244999b0f52464dc443 : SUCCESS in 18m 38s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/882c27d0b86744afbf3e02a883b319d2 : SUCCESS in 25m 10s","accounts_in_message":[],"_revision_number":2},{"id":"a8be57989dbadb97009f81b49d6fa978c69f1560","author":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"date":"2020-06-18 10:23:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"61348d5872347b9728dc6d8caf484ab99a619f4a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 10:26:10.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"26d9e52c3f5261eac37ee9a6d9172dc1f5133c32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:35:09.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd918c66a57443349d2dc0aeab27fac0 : SUCCESS in 7m 09s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c49e92ba1f224789b0d5958c935b6dd5 : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/56272cdb34824a52a837a2309d5f0410 : SUCCESS in 7m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d14ddda7d544f71b0b9b917abe1cf3e : SUCCESS in 9m 09s\n- kayobe-tox-ansible-syntax https://zuul.opendev.org/t/openstack/build/4e784765712c4450a75e0516304c5327 : SUCCESS in 6m 47s\n- kayobe-tox-ansible https://zuul.opendev.org/t/openstack/build/dec6b84ca663457ea06bb375cd360dbe : SUCCESS in 12m 25s\n- kayobe-tox-molecule https://zuul.opendev.org/t/openstack/build/6395ec0e3a6e4693b22c1697db30a74a : SUCCESS in 36m 29s\n- kayobe-overcloud-centos8 https://zuul.opendev.org/t/openstack/build/10f8ab030c0d4d1cbcbe3288ad6183a4 : SUCCESS in 1h 15m 01s\n- kayobe-overcloud-host-configure-centos8 https://zuul.opendev.org/t/openstack/build/2558ed829f5949caa5b3768a819520db : SUCCESS in 21m 26s\n- kayobe-overcloud-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/d155b2c869d24508902faf95bb9f244c : SUCCESS in 1h 50m 24s\n- kayobe-seed-centos8 https://zuul.opendev.org/t/openstack/build/c9d8b306c7a647eeaab55ade973ad0ce : SUCCESS in 20m 36s\n- kayobe-seed-upgrade-centos8 https://zuul.opendev.org/t/openstack/build/b0a6b8e6644b42f99dcbd1923cd481f5 : SUCCESS in 30m 42s","accounts_in_message":[],"_revision_number":2},{"id":"ffd09864b347b516d732f1271cbb6a24d57e2b39","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:35:09.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"55363c2a15aa703e39e8e2590178a118f6da37d2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-18 12:38:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36be2b038dad40ed94b159882408d44a : SUCCESS in 1m 57s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"4c0bd3c93b2e0f285dea3552545f7bfcbb05446c","revisions":{"6f6dff9dce1918bec5a8f0e3495f4e8bd0281f06":{"kind":"REWORK","_number":1,"created":"2020-06-18 07:53:23.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/736484/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/736484/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/736484/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/736484/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/736484/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/736484/1"}}},"commit":{"parents":[{"commit":"e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e","subject":"Merge \"Add framework to deploy user-defined containers on seed\" into stable/ussuri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/e9397b4e4cca77b6e0daf0e51a1eb598badd9d6e"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-06-11 15:38:38.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 07:53:23.000000000","tz":0},"subject":"Update image regexes","message":"Update image regexes\n\nSome images were not being built automatically due to a missing regex.\nThe most notable being chrony.\n\nChange-Id: Ia0aa15300f58fc4e912dea7ed304a1cdaf51382a\nStory: 2007796\nTask: 40050\n(cherry picked from commit 50317285cba35ee919c5fbc42249dfd6f072f14e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f6dff9dce1918bec5a8f0e3495f4e8bd0281f06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/6f6dff9dce1918bec5a8f0e3495f4e8bd0281f06"}]},"branch":"refs/heads/stable/ussuri"},"4c0bd3c93b2e0f285dea3552545f7bfcbb05446c":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-06-18 07:53:38.000000000","uploader":{"_account_id":14826,"name":"Mark Goddard","email":"markgoddard86@gmail.com","username":"mgoddard"},"ref":"refs/changes/84/736484/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/kayobe","ref":"refs/changes/84/736484/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/736484/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/736484/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/kayobe refs/changes/84/736484/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/kayobe refs/changes/84/736484/2"}}},"commit":{"parents":[{"commit":"842d4b999df4a043e433e85594d725bedbb9b6cb","subject":"Update feature flag defaults","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/842d4b999df4a043e433e85594d725bedbb9b6cb"}]}],"author":{"name":"Will Szumski","email":"will@stackhpc.com","date":"2020-06-11 15:38:38.000000000","tz":60},"committer":{"name":"Mark Goddard","email":"mark@stackhpc.com","date":"2020-06-18 07:53:38.000000000","tz":0},"subject":"Update image regexes","message":"Update image regexes\n\nSome images were not being built automatically due to a missing regex.\nThe most notable being chrony.\n\nChange-Id: Ia0aa15300f58fc4e912dea7ed304a1cdaf51382a\nStory: 2007796\nTask: 40050\n(cherry picked from commit 50317285cba35ee919c5fbc42249dfd6f072f14e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c0bd3c93b2e0f285dea3552545f7bfcbb05446c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/kayobe/commit/4c0bd3c93b2e0f285dea3552545f7bfcbb05446c"}]},"branch":"refs/heads/stable/ussuri"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
